)]}'
{"id":"openstack%2Fkolla~694235","triplet_id":"openstack%2Fkolla~master~I26ab8824d8c6c108eadb85698dc8367b576f485c","project":"openstack/kolla","branch":"master","topic":"no-py27-jobs","hashtags":[],"change_id":"I26ab8824d8c6c108eadb85698dc8367b576f485c","subject":"Drop openstack-tox-py27 job","status":"MERGED","created":"2019-11-14 10:31:03.000000000","updated":"2019-11-16 12:21:30.000000000","submitted":"2019-11-16 12:19:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694235-1573906799506-1353a74c","meta_rev_id":"32456d0c58f91ccc45aa67df60b05a608f3c113a","_number":694235,"virtual_id_number":694235,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-16 10:30:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-16 12:19:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-14 10:43:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-14 10:42:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-14 12:53:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 10:42:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-14 12:53:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-16 12:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d0b3d4aab592c06e8dfb2446ac9a0ede17580e4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-14 10:31:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda78fb4c05d545ea20936bf131fbf31ad8d3867","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-14 10:42:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd0d08dda8485e877010a655adc13d8d74154984","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 10:43:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b029b5b44cdf553151c4e8d190bedc342ce311c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f8cd4571e16f4a3da0647910b2f769ea : SUCCESS in 34m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/358811957a724eebba9feed36de0b33d : SUCCESS in 1h 03m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c273e45d9996481e8eaa4b7e359a762c : SUCCESS in 1h 27m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f23ecc738789457f98df64fe3f8e43ca : SUCCESS in 1h 27m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e986c7d9fdc9434ba794f269a987b3c9 : SUCCESS in 1h 50m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/68d4481a357b4cb8b2ab808f1b42005a : SUCCESS in 30m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6fc7a9b67abb4b76818110f8b64ce70c : SUCCESS in 17m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed7117a20f434a7ab336582a6a51090f : FAILURE in 21m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c04e07053a9b418a8d1eeea59475cd94 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d85f24b8e844588a99e90d66f4219e : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8750850abac74a55906a52af7196687a : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88bbb9b3a23348479ab8121c6e549ea0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc8f7ae4d2e480581b30d9d2363a6f6 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aeeb739df929493fa6552a36cb1ed693 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a6c2d3bc5be4b05845a36086b118f39 : SUCCESS in 39m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7df46c9bf5f84b9795a0007852af29af : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d597f4350f28420dba2c84a06c88904b : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/900c16c6bba74a659081db35b9e5bddf : SUCCESS in 1h 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9c5c9a9dd9eb8e83d29271ce3c4494b208562452","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 12:53:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce5a31d70aa387ad273614445823395bf1d0276d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:54:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6d8a16bba1470a381653dc7b5a17e0cec4d38bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:44:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7fe8b73ad7134724978d122c03dfb045 : SUCCESS in 45m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/867af28d02b24c559868f6243505275b : SUCCESS in 1h 10m 23s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4b58e7f3e68e41ce9c975200bdbd07cf : SUCCESS in 1h 47m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae97989d6dbe4a4580ba834a80a5f707 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef582ae7f0d24c679c03d1c44a4fee74 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02adc04dbf6f414fb517d3ec210d0543 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a622c348d3354d5cbef98551b462cc58 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3084f1cf2ad4b339628737c7e582c12 : SUCCESS in 9m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b9769d2640a404a8ee7a159d88fcafc : FAILURE in 5m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4637e5ee7724efcb43bbddfb8363361 : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85d0a054aea8448a9882a783831a5c88 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":1},{"id":"736e18856a80c3915dacd08ff8f939e609e61413","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-16 10:30:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6baa359ece19f041e6dd8bc04f650e6a1ede5509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 11:26:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ea2cd4579adc43bfb4028102804c6a21 : SUCCESS in 45m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/993bececb1b644329180a54d4030bbf1 : SUCCESS in 54m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/679ed3d9987546b08a23298d2cd376a3 : FAILURE in 12m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd01f8aa25ba4597af530d26e46c3b96 : FAILURE in 13m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae62ae6a2e044cb2836b2a518c21a7f8 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/550f5e51a73c40b1a0c87790a830504a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b7235720be74592ad7648a86873012e : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0fd7d7f2077480e97685eab0d595d1f : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac49883a5de4647b8f17c9cee30ed6c : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd679d1d36334b8a9b2e6fa29f9a2f92 : SUCCESS in 31m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68dc0c6635964bf5ae83bcdf0647d6b4 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c8bfc40df14556591e8a123ec237211046c9547a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 11:26:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8dc3ef0182391bc77c6a613677214b754be7e2fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 12:19:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ccf373fb2a2c7e998a49015b7bd1587cbf1b7e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 12:19:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/84b69908c29d49eab1e16ebbf72b70f4 : SUCCESS in 48m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed6a4fff8a648cc800bb320e4b9239e : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fdf34d537e447694de6efe9a27e440 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eca980c52359448da7ab08219354ca4c : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d4f382e881b4dc68004fea7711da5ad : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52602104428347fab4592301d61d1025 : SUCCESS in 4m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9edf1c951924fb7a08a83bf768a5cd3 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e05e9fc3c5c395cd38d46fa2bec160c4808a7d99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 12:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f802b52dbe7e4bdbbf7030a709752030 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c9106bcc18b7bc92074bd4afeb091840843c0c1","revisions":{"9c9106bcc18b7bc92074bd4afeb091840843c0c1":{"kind":"REWORK","_number":1,"created":"2019-11-14 10:31:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/35/694235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/694235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/694235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/694235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/694235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/694235/1"}}},"commit":{"parents":[{"commit":"414987b15f7d79cf4c51f305589b5b6d2d8fe1b4","subject":"Merge \"Remove deprecated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/414987b15f7d79cf4c51f305589b5b6d2d8fe1b4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 10:30:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-14 10:30:38.000000000","tz":60},"subject":"Drop openstack-tox-py27 job","message":"Drop openstack-tox-py27 job\n\nChange-Id: I26ab8824d8c6c108eadb85698dc8367b576f485c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c9106bcc18b7bc92074bd4afeb091840843c0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c9106bcc18b7bc92074bd4afeb091840843c0c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
