)]}'
{"id":"openstack%2Fkolla~694453","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I31a372d801dace92950cd85b29b308431f78bbdb","project":"openstack/kolla","branch":"stable/train","topic":"train","hashtags":[],"change_id":"I31a372d801dace92950cd85b29b308431f78bbdb","subject":"Use ovn-ctl script to start OVN ovsdb-servers","status":"MERGED","created":"2019-11-15 05:33:26.000000000","updated":"2019-12-03 15:02:24.000000000","submitted":"2019-12-03 15:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694453-1575385203962-232123cc","meta_rev_id":"24142198e7ae34acdfaa23711e0bb37ee0cd4d3a","_number":694453,"virtual_id_number":694453,"owner":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 15:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-18 15:50:51.000000000","_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-11-25 14:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-18 15:43:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-11-27 10:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-12-03 10:13:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-11-25 13:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":1,"date":"2019-12-03 10:13:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 15:43:58.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-11-25 13:59:30.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-11-25 14:14:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-27 10:38:42.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-12-03 10:13:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-03 15:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bc2d0c110db25cf78c2f52afdbc3d3429083098","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2019-11-15 05:33:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3996f5b73e5bda0340485f71f2a54e8228173f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 07:02:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/666414cbfcc741fd9bca32ed3b9b10b8 : FAILURE in 2m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8a847a7d5df4a78adcae5696001f8aa : FAILURE in 2m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8b563fb461464599982dec3eb1363071 : SUCCESS in 1h 24m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ed79ee43d07840a1be12eaa0bcd2bd04 : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/122464cf619b44f6823f586c9d66cf2d : SUCCESS in 1h 20m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/99034ab0d7a441d1a2bd0d58cd70ad1e : SUCCESS in 56m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/36a3ff660c4d4c6183dd2c3a45940a0a : SUCCESS in 28m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1313e18ad55f458b8670cdf03a647812 : SUCCESS in 18m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/966e3bc088ea4253b47e20b45dcd6de8 : SUCCESS in 50m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7838993e5aa040aa96cc0afa24164723 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc3fe2843754f3494a94c00257c0e53 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ba6a9a5999e449c9fb9484280b9bef7 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/965bb59558524c8794f5fd8dc729ece0 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b25c9c9f5528416e93196c828df806e3 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac0bd430c7d498d868901b267cdc039 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/345fc1a229134795ad4fa72651ee958e : SUCCESS in 41m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7414367e9c641c2bd88e0fc51558dff : SUCCESS in 29m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70948bbdd28d45e785167593b16eacbd : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/396fb12885684a21ae01c3bdb29ff77f : SUCCESS in 1h 06m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b88343673fa4ab389965fbb963a6367 : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"832c55ff86daf300bb0665151b850a409fc03641","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2019-11-15 11:10:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"02669bf84c27a03aa5fecd37acc3307e43cf795a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:44:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e739e4f706d14398853bcb863d388dd0 : SUCCESS in 1h 16m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bd58108201494e2ca89c30793546eee3 : SUCCESS in 1h 22m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c8cd34b113d34140af23c3f236527a4c : SUCCESS in 1h 46m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59457ae0cc6e48ec97f0930d00dc4201 : SUCCESS in 1h 37m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d8c9146824284fe4b4a184575c7ef6aa : TIMED_OUT in 2h 32m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/94543d952ef94e5d9eaac106908d1eb4 : SUCCESS in 53m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6ac632024394cb7be6dac136444a6cf : SUCCESS in 27m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1cbaa06b19224d61b661f5d192eec427 : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6d5d8910db14421a624692061d5c960 : SUCCESS in 50m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecee431a06b743228d0733b1b214c80c : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9e9d055f9b43b3a7466c235a943155 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/310682af1da047bfb20f804dc0249377 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5aaabb95dc042f7a6585f797393f691 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/495934f0acf6402197ff498a6219c187 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a6f4b57dab4927b6045e601a06c1d4 : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b07bb621d1e943bf8d48173b7bab43f8 : SUCCESS in 27m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6d66958bba64617ac55ad93ed7e3b24 : SUCCESS in 29m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20c5209a271647c5b6f4aff598d37612 : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eaeb7ebddd6c4e509f22ccd9a31bbcf3 : SUCCESS in 1h 11m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c29a0bc637b4a7785ddecb5b128f2ca : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"042255c2d71e233754ddf4b81c2aa039d9128bbe","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-18 15:43:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03a4a2c17212ffb396e06cf5adbadfb14a869fac","author":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"date":"2019-11-18 15:50:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f8ff7159f84bb450d01bd0c3fa69d24705b5c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:27:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1695e1d6f37d4976b1e8c5d8e9fabb52 : SUCCESS in 49m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5b79c3f3e0d54e038e72643bb365d9e1 : SUCCESS in 48m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ba7a5166534d42ecb83b8bbcbbc7b3b2 : FAILURE in 56m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5112aac6677845d9ae8ccdb06181f74e : SUCCESS in 1h 30m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d3f2a12f4279433793dc17dbc1a736b2 : SUCCESS in 1h 32m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f0b014c1810f4ed9b414fa50e87afbdf : SUCCESS in 55m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c045f0bbfe9b4582a3c532e55bd4844f : SUCCESS in 31m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/788f7e25cf564a9083344854d53d0aca : SUCCESS in 24m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/903f42ce271047dd88cfe718675c42c7 : SUCCESS in 59m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62010ddbee324539ab0f975caace5099 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6e42d1ed234b259527b5d0f59cf36a : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b276e5a236e470da7c9941de6c5df01 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56eb9188ad4e4205b79f0ffbea9a19af : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e07b8404c5a4fca97b31ebdd5763e8e : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d54ad6a3324a1390fa1809a78f8aac : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/855a300e934e45cca5d970aa61d74bfe : FAILURE in 36m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3baaf4e840bc4f20921fd5895fe14f25 : FAILURE in 34m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d68488c2a3e9435fa9f3cb959d32aa8d : FAILURE in 24m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0a843b09b024b39be1d24fe2d2db6d3 : FAILURE in 22m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ce7273c692b42198ee4d5cf1f41b630 : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d7b25d5dd594c80a9763a8f1315659704dd1eb2f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-25 12:04:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7abd710ed6babd64efdaed0314e02870bcc6c06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:41:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2080fab8b087442798107c7768d97c20 : SUCCESS in 38m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/36143d29df1a40fc88f5b338eacc0b3e : SUCCESS in 54m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/736f7aa9ed5c4f77b72bcad5b5825a4b : SUCCESS in 1h 26m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8124557de8a4435dbf04ada657b8fe44 : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/29d6c4d6be0a46a2b626291fe9917982 : SUCCESS in 1h 33m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5365ec514ba049b3a0f7ce4957475b20 : SUCCESS in 57m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8a63f92afa674a60b96e7dc92ff0ce99 : SUCCESS in 26m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d3fc70f7b16412ab232f02caec20f81 : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/88f03c3bc4884c9a8a9dc9b47355b47f : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccac37b3f3744ae4a9ab6a192ad7b725 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c60ddf1c0df423a986c96bde107af91 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17e002e056d24548ba0cb3a160c92b86 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78c2aa70cfa241daa909c15ef39ffdbf : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba7e05dd065a495dbafd4aadfb8606f9 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf357f7999c84265aa73d3220e684c55 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/848f3378c6604a03bdc4dbc52ede3a09 : SUCCESS in 52m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d57b825f90b4710921de75cd82ed679 : SUCCESS in 56m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a055820c25c742ea9264a8de43eb9854 : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4439d7e1c2574d67b18e5c77961ddc71 : SUCCESS in 55m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bb37cb29d774e4d9cbb482a0fbe797c : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b37ffb809d21a7ef96bf92c499efbc203458eb3c","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-11-25 13:59:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d52f7b45cc837bdcf0b1e80fa015d143fbb08607","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-25 14:14:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07d7d3302cd599ce6e63646bb663c1c67921fd6a","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-11-27 10:38:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e2a75a957bf5f6f9576fad431ee4ae472f7414d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-03 10:13:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ffd9d24f615cb4d152f6a09e6a01dbddc881371","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 10:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2316cc066ffa77d89be613ede34a36fa06d1f5e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 11:39:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a640fdeec0474854aae53e5e9de2b978 : FAILURE in 8m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7d216e718ef54aa38c2109977718a089 : SUCCESS in 1h 21m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/308c22406f8b4198915d1a59efaf21b8 : SUCCESS in 1h 20m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cb414f1497854b27997e7f94ea8a2a41 : SUCCESS in 57m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c7260c478934a12aae5b352412752d0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24bb8f9f180a442b8912da06977079b0 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e95c4ddc5014fd896c0f4e3c585f67e : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e45caf8566e849c2972d47b48887e167 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d2e538fcc704e73aa71656057facd3a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26afb7e9b13d43f79fbb5d72db95fc7b : SUCCESS in 6m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9471ce9564ce4dcda9c4e27b159f2165 : FAILURE in 5m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3cdec39aa4e4250b6fced3f98be2cfe : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70d32d6e72d04091bfec1acbd51612a3 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d5b54d5a54c1ac40d8359e072bec1b5c79848284","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-12-03 11:48:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60b8d960b6e703650458c9580a63a740e398c9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 13:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a67362a3256b4a969442e91600e2eab0 : SUCCESS in 49m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/57b9a168ea3245d8a8696b6a455c398b : SUCCESS in 1h 08m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e084e698372416b968f87ffa8932fea : SUCCESS in 1h 33m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b1801553eed94614a1b2362d55bca0da : SUCCESS in 1h 29m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f582d5012f4b4d4184e1f1c22327b522 : SUCCESS in 1h 35m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/20d7cc9cf6794d45abaa226ced90226b : SUCCESS in 46m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31b0b0285dc24032b2a8c965c1180e3f : SUCCESS in 23m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8727266bff34b4dbbe5a3e39c7b5cc0 : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/46c540a2bb144736acca7b8f1e51b38f : SUCCESS in 1h 10m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da6a9841c4e442cd894dd0f887512890 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c8204fb3534ee680c023a4daff7eba : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3bc121ba0754855879f2ed5f8299a62 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91231f3d16414c70aca88ef3c3f79d70 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7b3a97592ab4ecda5f7074baed40a0c : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ba1b62894d4270ad905e6a69fa2e8a : SUCCESS in 7m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e159b97449143c7ba7a204cace4d716 : SUCCESS in 26m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/350ea6c724304912854d6053a909cdf5 : SUCCESS in 37m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23ef8bef082f4fd0ac8540c143e288b0 : SUCCESS in 1h 08m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e5a44cf31f44621b312c9d964302ba6 : SUCCESS in 1h 15m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29049293c91447d9a9cf2545e088fee1 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"02bacd111dd60c0e0c8f5a93a50bbf79e0d17dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 13:30:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73d54cf018fcf7b74b1337e18f8935ad2815e0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:00:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7079a897391a464388688279a5e0e4b0 : SUCCESS in 46m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8d73c4c2665045168d68866b3147a29d : SUCCESS in 1h 24m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cebae16a1033438885078e1a2b32212a : SUCCESS in 1h 27m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fae2bf65a32a4facb9e2ebda4bb46d6e : SUCCESS in 56m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41a2ff05f9d849c3b2671c5b6d483aef : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a6e29f40544f8794afd4699b331f99 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd8e42ff90b241599376fba033c0704e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5fe03e4438340c7a60db0159beacbd2 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa076c9bc22d48c69046dc077e6314af : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b5ecd02d694254a0361f82eb697092 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2945aaa3565444b1b2bb1deeba60af69 : SUCCESS in 38m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cf14602d5a947e9a43cfd5a82ca59c7 : SUCCESS in 56m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b6a9a28e0c8d4ba2857d8cc8271310c6 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1ad3194d74a0811a2b1c3fd98cf0d7e1895ae04f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:00:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e983f62a42a18a99f15612c1fa1a4b385420d363","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:02:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631c06db19a64c38aee54090ca4be2d1 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03668df367d3ed2aa82f6310f746d958a8a15f7e","revisions":{"03668df367d3ed2aa82f6310f746d958a8a15f7e":{"kind":"REWORK","_number":1,"created":"2019-11-15 05:33:26.000000000","uploader":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"ref":"refs/changes/53/694453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/694453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/694453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/694453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/694453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/694453/1"}}},"commit":{"parents":[{"commit":"f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1"}]}],"author":{"name":"Numan Siddique","email":"nusiddiq@redhat.com","date":"2019-10-27 07:05:47.000000000","tz":330},"committer":{"name":"Numan Siddique","email":"nusiddiq@redhat.com","date":"2019-11-15 05:31:15.000000000","tz":330},"subject":"Use ovn-ctl script to start OVN ovsdb-servers","message":"Use ovn-ctl script to start OVN ovsdb-servers\n\nUsing the ovn-ctl script to start the OVN ovsdb-server is the better\nway. This script is used to start and stop all OVN systemd services.\nThis script takes care of creating the database from the schema,\nupdating the database from the schema when the schema version gets\nupdated.\n\nThe present scripts - start_nb_db_server.sh and start_sb_db_server.sh\ndoesn\u0027t take care of schema updates.\n\nWith this, we don\u0027t need extend_start.sh in ovn-nb-db-server/ovn-sb-db-server\nto create the database from the schema.\n\nThis patch also deletes kolla_ovn_extend_start from the ovn-base image as it\nis not required.\n\nCloses-bug: 1850025\nChange-Id: I31a372d801dace92950cd85b29b308431f78bbdb\n(cherry picked from commit b8c8fa15edcf5336a1121155cdc880a35ccf161f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03668df367d3ed2aa82f6310f746d958a8a15f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03668df367d3ed2aa82f6310f746d958a8a15f7e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
