)]}'
{"id":"openstack%2Fkolla~695202","triplet_id":"openstack%2Fkolla~master~I66b4c93d8e6f50086a014b9c828fc98943686837","project":"openstack/kolla","branch":"master","topic":"fix_skydive_source","hashtags":[],"change_id":"I66b4c93d8e6f50086a014b9c828fc98943686837","subject":"Skydive - make source images download from github releases","status":"ABANDONED","created":"2019-11-20 14:22:36.000000000","updated":"2019-11-22 14:35:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fff62735eced8b21a0c7f1e76474a4786b8e33c3","_number":695202,"virtual_id_number":695202,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-22 14:35:16.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"date":"2019-11-22 14:01:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-21 10:35:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-21 09:32:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 15:18:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-21 09:32:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-21 10:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-22 14:01:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"928e4181304807bd3a3f02eac32fa4fc26f576bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-20 14:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ad62f32bff5b7bfd66f8998c66b8d155c144f8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-20 14:54:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/02/695202/1/openstack-check/tripleo-build-containers-rhel-8-master/f68ba35/ : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"90a9ee55888b765a2326b8c4041d198036b77054","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-20 15:18:42.000000000","message":"Patch Set 1:\n\nnot sure if \u0027fix\u0027 is the right word, more like \u0027change\u0027 or \u0027sourceify\u0027","accounts_in_message":[],"_revision_number":1},{"id":"0a415b136ddd4b1c3cb0463d9b3a5041c89bf590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 16:00:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f045ca2e3fd74bcbab8bc65e196c9843 : SUCCESS in 48m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/82001bba61b8402fa7052acb41550755 : SUCCESS in 56m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e75e09d0b07f421d9f9376f2e4d00138 : SUCCESS in 1h 15m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e97a9a6a71d141e59d7b374adbd7f0da : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/20e1851930734cf585f9838432a9e2a4 : SUCCESS in 1h 30m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3b7f09cb6a374845ac44eb3e0b795042 : SUCCESS in 1h 02m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/22d43699159849e192cb62e6c38d2f07 : SUCCESS in 25m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/81c746b177df4b1bac63d9d6dd3135d0 : FAILURE in 16m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74e7e2d650d74e03908b58bddd49e2eb : SUCCESS in 57m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f8f7a3997f46e6b0ac25847ea08c87 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ad43b8ce6742deb2f70b5c92f32c69 : FAILURE in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5de25b04e12744cb9469090c114d3df9 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2964da37b2944dffaa4a99755356b08e : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9c08c31ce247e7bb585dcaf6360691 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e61a7f293ef245f783c89caa02252e93 : FAILURE in 26m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2d84ba7b3c0487aa4dd7d9cc1bd861e : FAILURE in 23m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5dc5a4e53ca4aaab9cf20ee4d29d80f : FAILURE in 11m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7227acf09cc4e47a5a2f5b627a244ed : FAILURE in 18m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03dca21821524759b83951cf25f28f32 : FAILURE in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"758385b508ea55321639a0e7b69b0dde274f74a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-21 09:01:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"067e5acb9ad115d02fbba1ac2b84292e3db28722","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 09:32:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/02/695202/2/openstack-check/tripleo-build-containers-rhel-8-master/b69816f/ : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9ca8247e1030f0ca6840a0c9551a034dcad11bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:35:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/17f921cf432f4e3896e4dc5b6fc0b60d : SUCCESS in 38m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/243743a854fe479080995ad43c41addb : SUCCESS in 52m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2e6696ecfc034986a6e60736dadbcdc0 : SUCCESS in 1h 33m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ac54f4725afb43f4bcbfa967b0f64a0e : SUCCESS in 1h 14m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ea13312b283b45f88c0e745ecc111a1b : SUCCESS in 1h 20m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a19be42dd2464d419168984c9bc63f63 : SUCCESS in 48m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9188d986e1e48fb9df456f3145cb7eb : SUCCESS in 22m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45fe7bc7839f466d95ac26ddda14a077 : FAILURE in 30m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/699ec0bd73bd48479c1b78f01bd47604 : SUCCESS in 1h 10m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02a9d634f5e74e5db97151708b1c7bc2 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f296d623d914f89bb2945671127f24d : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424290893f6941d4a75315f6bf615320 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/723d9b3dff964079b99ce7c1a1b63279 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1d5cddeb984181992bd22e02783dd0 : SUCCESS in 8m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b477b24b85284cbe8c579f9ba7bdc7bd : FAILURE in 32m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a706666a7fe486da01111782cf1d3cf : SUCCESS in 35m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53967d03df134da6b73a3b62227d3232 : SUCCESS in 1h 02m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef5d8b3314184076bb0fbddd9fcc3363 : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e5d0e6b6a40b44dd9219ff523e40ca96 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5c22019860ab64492ee109caeb4c26aed8c7905c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-22 14:01:21.000000000","message":"Patch Set 2:\n\nThis doesn\u0027t really align with our new \u0027infra\u0027 image type. :/","accounts_in_message":[],"_revision_number":2},{"id":"c2634feda702a9971eab1dcb50b4aaf8cc93d9db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-22 14:35:16.000000000","message":"Patch Set 2:\n\n\u003e This doesn\u0027t really align with our new \u0027infra\u0027 image type. :/\n\nSure - I forgot totally about that - I\u0027ll abandon that, and we can start preparing for doing \u0027infra\u0027 image types - I hope there we can use whatever is better - rpm from distro or code directly from GitHub.","accounts_in_message":[],"_revision_number":2},{"id":"1e4048e65d365b753c9902dbec647550f5b6edbc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-22 14:35:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"692b163bcaf0023f1a705a9075c73f2d6fea67a2","revisions":{"eb615af7a6e6f8b2852b116d5b6c21a3803b37f4":{"kind":"REWORK","_number":1,"created":"2019-11-20 14:22:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/695202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/695202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/695202/1"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-20 14:18:53.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-20 14:22:34.000000000","tz":0},"subject":"Fix skydive source images","message":"Fix skydive source images\n\nUntil now - centos-source and centos-binary did install skydive rpm\npackages from centos-opstools.\nSince centos-opstools stopped updating their version of skydive package\nat 0.22.0 (current version is 0.26.0) this patch changes the centos-source\nimages to download the binary from github.\n\nChange-Id: I66b4c93d8e6f50086a014b9c828fc98943686837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb615af7a6e6f8b2852b116d5b6c21a3803b37f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb615af7a6e6f8b2852b116d5b6c21a3803b37f4"}]},"branch":"refs/heads/master"},"692b163bcaf0023f1a705a9075c73f2d6fea67a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-21 09:01:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/695202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/695202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/695202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/695202/2"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-20 14:18:53.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-21 09:01:05.000000000","tz":0},"subject":"Skydive - make source images download from github releases","message":"Skydive - make source images download from github releases\n\nUntil now - centos-source and centos-binary did install skydive rpm\npackages from centos-opstools.\nSince centos-opstools stopped updating their version of skydive package\nat 0.22.0 (current version is 0.26.0) this patch changes the centos-source\nimages to download the binary from github.\n\nChange-Id: I66b4c93d8e6f50086a014b9c828fc98943686837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/692b163bcaf0023f1a705a9075c73f2d6fea67a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/692b163bcaf0023f1a705a9075c73f2d6fea67a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
