)]}'
{"id":"openstack%2Fkolla~695356","triplet_id":"openstack%2Fkolla~master~Ic74fe3021154826c7db577ff8d7008bfef2ccd1a","project":"openstack/kolla","branch":"master","topic":"bug/1853167","hashtags":[],"change_id":"Ic74fe3021154826c7db577ff8d7008bfef2ccd1a","subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","status":"ABANDONED","created":"2019-11-21 07:09:08.000000000","updated":"2019-12-05 19:32:29.000000000","total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a20311b4ddf65d2ef5ec6978718ff8e6c62e95d","_number":695356,"virtual_id_number":695356,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"date":"2019-12-04 14:59:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-02 17:15:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-02 15:16:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2019-12-03 09:33:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-12-04 15:32:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-12-03 09:33:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-12-04 15:32:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":1,"date":"2019-12-05 19:32:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 06:52:14.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-11-26 20:30:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-29 11:20:30.000000000","updated_by":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"reviewer":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"state":"REVIEWER"},{"updated":"2019-12-02 14:41:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-02 15:16:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-02 17:15:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 09:33:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-04 14:59:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-04 15:32:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"adb36cdcddbc1c1fd2df9a17e445fae682978d66","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-21 07:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caacf94705ce99f6543b27591fcddd806a5407c7","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-21 07:10:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1522dd9c88ad28e9767466ba36b0934c405c86c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 07:28:35.000000000","message":"Patch Set 2:\n\nfor ubuntu it is not supported either - could as well streamline both with 21; also might make sense to pin rmq to 3.7.* instead of specific patch version","accounts_in_message":[],"_revision_number":2},{"id":"8eecdf359dbc156aeccf6e90666594714dd96c83","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 07:40:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/2/openstack-check/tripleo-build-containers-rhel-8-master/7e75bea/ : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1ae99ec75cbb0947d4311481aabd1136f0236f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 09:35:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dc17eda652f64fe8bcdcc166bf03d111 : SUCCESS in 48m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/57c5eb204bc2449c99002f18b2fac431 : SUCCESS in 51m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/723b86366c8c421da73393108a1ee5d8 : SUCCESS in 1h 25m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bb6a5267fddf41d9b16c8b611d19febc : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1fa49f02c1fa4ab3ac032e8c206453c1 : SUCCESS in 1h 43m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9578ba66c5504d85bd44145c8b4141f9 : SUCCESS in 1h 03m 11s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/8c06db3122cb4beda03c26050bf6a47e : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/54f17f38025a45c6b6f5cb1be56d0be6 : SUCCESS in 2h 09m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/62a57d403a6a4531886a8a4d9f311b03 : SUCCESS in 2h 19m 15s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/5b37bceada8a4e6fa76aff215100dda8 : SUCCESS in 1h 52m 22s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/32fb7d94f6d0455294573db1634928e9 : TIMED_OUT in 2h 02m 46s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5b92918570814a99969bc65016aba6d2 : SUCCESS in 1h 50m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8c159787908040f3b9f4d605d5fe1b09 : SUCCESS in 25m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0b4a6671213543c0958e189aa232ecab : SUCCESS in 19m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/061deab8db7a4f30b342339889f2f402 : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c100b6c57054293a3df97a3143b10fd : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb377f8b719422eaade87dd8a507cc4 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c520ca5570b4a4bae005de07840a649 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b655307a0104568af1f160ca232db96 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5709ad6c38cb419ebf42490f87523007 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71df60cf31e04ccea7a66f9445faa87e : SUCCESS in 22m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d7c0c9a88ae42b289d2ca42b21f5f17 : SUCCESS in 33m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fed309bd693d49c3b439c51220839cfa : SUCCESS in 55m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f88a4e69c1a646589c6e1e9b32ac8d07 : SUCCESS in 55m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9599f93494654ae99647eda246a53161 : SUCCESS in 1h 09m 48s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/83c3b9fffb6d4f2f86578e1d8d834242 : SUCCESS in 1h 09m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/256a5a2fec2a4e29a2227f5cb29cb7f8 : SUCCESS in 1h 27m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/36066f0ca94e410fb275bd0c11daeb94 : SUCCESS in 1h 35m 34s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b59cc68230c54e4dba05395944780c9f : SUCCESS in 1h 19m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/651a8129597f4012a2127e1d1075fbba : SUCCESS in 59m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fabc6f04db2292d8e66b8d86ba81e96a297dbf5","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-21 13:39:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"139d5cf9e35bac0ad17ea18fc9d4b7dd75cb5201","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 14:12:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/3/openstack-check/tripleo-build-containers-rhel-8-master/a640b6f/ : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a709ce8ad72ef320f6c7f645845cc88ed2d6694e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 15:37:44.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b27b85c1c37c890a2f2c0a47630a1242fafbb2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 15:39:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea07ebb45ef8b605925347566494d22d9292c9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:22:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d7fa4e691fef4a5bbf42919e80f26b07 : SUCCESS in 53m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3b67a14270e94174894fe885273a4788 : SUCCESS in 1h 00m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1cad648cdbfb4bafbc0194a94d9c2014 : SUCCESS in 1h 19m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/77e602937677497f9e4dd63242a72330 : SUCCESS in 1h 27m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0fbe6471cf374057bd117712eec729b2 : SUCCESS in 1h 32m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/30a784b235fb4518a9d50e51017a82da : FAILURE in 1h 00m 20s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/34d75126b18044a9bbffc45f1a42aec2 : SUCCESS in 51m 57s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/19f1fdf3309a46479ad02ae3e047da8e : SUCCESS in 1h 58m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5eb41919cf3344d38c946348f18d8f75 : TIMED_OUT in 2h 34m 23s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7915e3cf88b24f82886b3141fc4b7b9c : FAILURE in 1h 44m 27s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/555d3f6b622e41dbbf14584a2fe9889f : TIMED_OUT in 2h 03m 34s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e2154cdbd39c4308913fd944654ca2c1 : SUCCESS in 1h 55m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c7f921570b0485e9e3ed4818dffbe82 : FAILURE in 6m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ece46d04ed740f0ae95da16864e381a : FAILURE in 19m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6edd3e7f065e474681dff5530c188938 : FAILURE in 21m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/775fd12aa7ce470c86c821111dce1e49 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70e461ac5b94bc7b0a5be1420c7b691 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d4e883d9e5c4ca0a63b7f8cb3a31bf9 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/504bb827f6a54101bc26943dbb254dea : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c3ff1126ba4573891a10c2538754fa : SUCCESS in 7m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d8349dfe4744e4e853ed3da10b2703c : SUCCESS in 36m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7e62e5c5b7340e592df6453a238f310 : SUCCESS in 29m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dd9fa19412d4d0dad7b0c8cb05f4c41 : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b04c0d934ef44a84a5610f4a6647d4ee : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1f993960b66448491303a03a2d56f52 : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/07aa63c9c0ea43008e01109f1b4c80c6 : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/595f6d3ee4544252a16e9c1f04012d84 : FAILURE in 23m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4602e4a235bd4e359fd5e3380826dcd9 : FAILURE in 23m 13s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/591f5b126efa48a6a463ac438db173b0 : FAILURE in 1h 11m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/88a9e781a6fb4c7da1a17f2c3fb7edb9 : SUCCESS in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eafd13479bfb2235f07cd2650e02abb236d20732","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-21 18:01:01.000000000","message":"Patch Set 3:\n\nWould be nice to have info which exactl rabbitmq is used and which erlang versions it supports.\n\nDebian has erlang 21.2.6 for example. Enough up to rabbitmq 3.7.18 - newer 3.7/3.8 versions need 21.3+ version.","accounts_in_message":[],"_revision_number":3},{"id":"e31c497621fef56690cc04dd9f14aa4010db15bc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-21 18:13:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c32d77655ada2c6cf24676c66c749a812bc4466d","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-21 19:20:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"363c33e55914269a9bb249950d78c35d3bcc042d","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-21 19:22:32.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"523526687517e3b284220d9ff54f0a44573076fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 20:00:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/5/openstack-check/tripleo-build-containers-rhel-8-master/4f134e7/ : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6b253f782918beeddaac5676f45345fa4517224e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:01:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7fa9be30840f40aa86a58c351a57bf87 : SUCCESS in 40m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/864754947d5d48b285db4edbef9a503b : SUCCESS in 1h 03m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f279611a5cca4343bbdd9be2a87c8b13 : SUCCESS in 1h 41m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/484cde9a625244e0913c554878a6d69c : SUCCESS in 1h 42m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f23d6367ee9436297f2296d6f0862bc : SUCCESS in 1h 47m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a6df15320b944c16b9c05947d29f983f : FAILURE in 59m 50s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d756f29170ea4ae5836d571d5a6f0b2c : SUCCESS in 1h 00m 37s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c59e425f5ca14cfeb6f546f05da63cb2 : SUCCESS in 2h 15m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b24a1f08ee2d473cb572f525894e65b3 : TIMED_OUT in 2h 32m 40s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0fa03cc516614c75acf97fb16efe716c : TIMED_OUT in 2h 03m 01s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6fc074de26ac4bd4aa25fdadafac0b8c : SUCCESS in 1h 46m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/2ffaa2e1664445658a2bfd88f6c1a48b : SUCCESS in 1h 27m 03s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3580c87194da47de857b2f363f1e3578 : SUCCESS in 27m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a074db866de45e9b7f10a987f536b6c : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa0e906443f54dec9fa75e7d1655a96d : FAILURE in 22m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a5b3c1968df48838bbc9586eba7d443 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b152de0fe8e4908976ed56b150ec8b6 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2430ec195deb43ce8f400c34e68861c7 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf5424fad3e64a929be9a653b26e3c72 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb70f7c0d69486698450208f69254d9 : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca0cae12aeee4c0186bcd309555408d1 : SUCCESS in 32m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/537ec3ae7fe442d88df69559e8b9d67c : SUCCESS in 28m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5eade726e57a4a16b2da899ab37af322 : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c83c2e24905b425fa742a64cdb7fcf3e : SUCCESS in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fe414d4b759547ab854ecd505af28f00 : SUCCESS in 1h 06m 22s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/23dec8470bcf4230b47aed4beea72578 : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a08275d730e8448a86cdb2566487c80d : SUCCESS in 1h 18m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/159f8385f5204de4ac9ecb8c63314eb7 : SUCCESS in 1h 35m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c02841a464d2401b9dad158deb475ad2 : SUCCESS in 1h 22m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9e198e56b9e440bf8033f07ce0da917a : SUCCESS in 58m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f765be9c739eb1325cb56530e6a78c488735be3a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-22 06:33:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e6a35bd57ee77d3ced8e8d19b00d3fb1289e9317","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 07:05:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/5/openstack-check/tripleo-build-containers-rhel-8-master/94101b3/ : SUCCESS in 31m 05s","accounts_in_message":[],"_revision_number":5},{"id":"a3362cdd67380eccda0b59a880d816ba843e3e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 09:25:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/75e88071e9af4556a3cf65a5e30e7e22 : SUCCESS in 51m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/af4686ff673f41a29ead2878da00659d : SUCCESS in 50m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/73bd80a6ccc042859935d1c3dc47330c : SUCCESS in 1h 19m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1325b920e1ba48ed928bebefbdefd211 : SUCCESS in 1h 23m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ff184c455792411fb61cf30492af5617 : SUCCESS in 1h 27m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f75c02327d574dfdb16c474f3eed9f55 : FAILURE in 50m 38s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/028a9ee74eb049b1bf698db8a6fbe7f0 : SUCCESS in 48m 43s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b1abbb448da147b491e8ec250b64209a : SUCCESS in 2h 06m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/71247fa61e0d435dbd4ab3ba8014306b : FAILURE in 2h 28m 13s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/769379db16dd48ad9bc23a18ff58ed9e : FAILURE in 47m 09s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a86ca917fe0142d2851842f91ac1fdd9 : SUCCESS in 1h 50m 25s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/51e661b5011f40c98de2d76670e47c09 : SUCCESS in 1h 31m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29cbdfdd71bc42deae717d2dc83c0a01 : SUCCESS in 27m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/86cdbbf9af4e47d299a9690413c3c25e : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/08ba601a31594701a0fcb0764cbb6fc5 : FAILURE in 19m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a52f711f7ad94033b8ef7e253cef46cd : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba8b4b50edc44eab55c9dad93773562 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/887698cb78334ceb999c136f0722e7ab : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b719241678545198fc743bb421f2ab7 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58108eb6d4e0489ca87edfa839d081d3 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aaa7e08fffe14f1586522d2aef33d906 : SUCCESS in 43m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/faf42e990a2649548a719d98bea29c21 : SUCCESS in 44m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8dd356d978a941e48c954f7805af98f1 : SUCCESS in 58m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c1a0607413b41f3bd7dfaa94d9839af : SUCCESS in 58m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ef224fe927b426aa731326bcb4d5dc3 : SUCCESS in 1h 06m 39s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d13e151075bc40bb9b8e34855137a6b1 : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c27d8fa3153748c9b6ec10271e3d7a2d : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/233b016ab1d44f3fa87bd5c1378e9683 : SUCCESS in 2h 09m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/13065d5021c84ef2a61e3d913391820d : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a49e52a0c1db46f0910c797c3b4dc9f9 : SUCCESS in 1h 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65e4a0512af43df064608a6e4420f9e0e20c3af2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-26 16:38:52.000000000","message":"Patch Set 5:\n\nping, missing aarch64 and tripleo","accounts_in_message":[],"_revision_number":5},{"id":"a2267a9fa98fdd4fbc74e4269dec2da2ed8aee7c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-26 20:30:46.000000000","message":"Patch Set 5: Code-Review-1\n\nI made comments over on https://review.opendev.org/#/c/695847/ on how to land this","accounts_in_message":[],"_revision_number":5},{"id":"66ee2866688f1cbce5d9544b04bc523bd98a73c2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-27 17:08:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f3f75ac30763e1b49ae8f943d76d8c5b24c32d26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-27 17:41:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/6/openstack-check/tripleo-build-containers-rhel-8-master/1e97fe7/ : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d0d0cab85342c1128215133b77b60e475ee43843","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-27 17:47:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"88d1308a43934a030ac04727bbc919c644ed87b9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-27 17:49:32.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"b443fd49f1c79c1a7ace3dfcad20cfb0fb11b533","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-27 18:22:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/8/openstack-check/tripleo-build-containers-rhel-8-master/a961749/ : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":8},{"id":"03fbf1119d97d70fdb10f74e3e7f8e267a9c1704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:56:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3add685159ec4264926d6d436c84c854 : SUCCESS in 33m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/460b552d648e4fe4b9287898a390ed7e : SUCCESS in 39m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/006b8db20ffa4375b8f70fe6d8deddb9 : SUCCESS in 1h 42m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2b73e889513543b09f88730646563019 : SUCCESS in 1h 22m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0458a72837e64078a934354541721b4c : SUCCESS in 1h 39m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d81e07fe80d94c648aa6cdbdfa9a5833 : FAILURE in 56m 37s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/fe39cc16166c488990eeb2845274dd56 : SUCCESS in 53m 42s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/eab7e4eb22534e83a4b26821f73d34ca : SUCCESS in 1h 55m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/df0dec34e35347209d1cef9777873241 : FAILURE in 1h 49m 47s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ff2cef0a4d924ecea4ddaeeff8818fe6 : SUCCESS in 1h 51m 37s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3bc6b63093be4538bd71f0af7d70803e : TIMED_OUT in 2h 02m 47s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/dc3d9134ce3a4b608f568493a2a920a1 : SUCCESS in 1h 39m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb6e9fa099ec402c82aa2be64bc5aa18 : SUCCESS in 30m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96a3a51f8d8743038b304fa0ac5b0cb0 : SUCCESS in 23m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4e69ce4dcdb1409b9bbcc694d8fe3a6b : SUCCESS in 1h 01m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef8a130473a42fa933d342f36f542cb : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00f95fe7faa4f479f8ad74a120f7730 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e82f9c4294441fa370801758164500 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b0072e8f25041a28e7bd4f72855f21d : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abb2853c5a84a75a17707c471e6ce0a : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77a98b040dfe4413827b7ce8dca2ec03 : SUCCESS in 33m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f234eb798a04849adc367bfd741a82c : SUCCESS in 38m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c52b2cf2a1c24a8c9ca24b72b5bd6b73 : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/880e9bd054464764aed7a25c38083863 : SUCCESS in 54m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/860a7527c2a74a90ac8ac9acd4229cd5 : SUCCESS in 1h 11m 33s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e5a4a8c2993f431c9227fb2f761a0c3e : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c64e63337a174950876cb0151e41da40 : SUCCESS in 1h 24m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0a2dd6384da147d0b8d76eb7cf93fcc6 : SUCCESS in 1h 42m 47s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f09d228f7daf4536b8cd2169080fe33c : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0e77bfde910c4f0d9d15f09ddda68ccb : SUCCESS in 1h 00m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ab2bfd97910ad359cc338786b1e430c661fc0f24","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:32:13.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f4a02668276c138ba6cac02680ad906da1fb2983","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:33:14.000000000","message":"Patch Set 9: Code-Review+1\n\naarch64 support for CentOS 7, Debian, Ubuntu.","accounts_in_message":[],"_revision_number":9},{"id":"87b906b5114e5aefb897f69db09caa46c07f61dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 14:06:35.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/9/openstack-check/tripleo-build-containers-rhel-8-master/b283330/ : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":9},{"id":"ae644565277848b2ced7e7bc2a68ac5e268e6926","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-28 14:52:36.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"aca6673a9b4fea1a6e7ad6ea9fc602d3c4405633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 14:52:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":10},{"id":"d65810057eccf09d9bde2c24eb98b35f79ffea34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 14:52:48.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":10},{"id":"dc723534170c335148e5f3c6545297d4449187d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 14:52:48.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":10},{"id":"33987bac1bce5a0bda731546f67721c7be12c788","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-28 15:07:20.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"8ad9c279858ea932b910ef6d4406344f392854f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 15:40:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/11/openstack-check/tripleo-build-containers-rhel-8-master/87756a9/ : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":11},{"id":"26e53824b78d1be1f3bafc944de38df9ea2a45fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:39:38.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8716d2241f54414c90cca819a07753eb : SUCCESS in 42m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f3a3d8b5fbc2488cacbcc384c18af7b3 : FAILURE in 36m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/79b0fec79ce8427ea60f8e0ca8264450 : FAILURE in 34m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/33bd0519383d41bda99a91bb7c4bde0c : SUCCESS in 1h 23m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7f72fd64963040c68caaece633cb1254 : FAILURE in 37m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/420d88542e014ca39c7dc5251647747b : FAILURE in 55m 53s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d6bb6ef78184465da0dd581048f3a369 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/88105057662349d28ea43f72ee010acd : FAILURE in 37m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/639ceff4b6dd4e2db1ea0ae0cd5eacd3 : FAILURE in 34m 18s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/12d4b12b7d184dfcabd9d8e63499ca22 : FAILURE in 31m 54s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4b073acdc17b4b38b2af98fff4846ffa : FAILURE in 35m 27s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/06a1e052962a4cc5a2006060811cbe5e : FAILURE in 40m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7032fc7f5a754c81a9e2939066820c5b : SUCCESS in 28m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/40eb884cd2fe41f89d9cd90f39130241 : SUCCESS in 28m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/acb0f084c56346489c53c3e0de694e93 : SUCCESS in 58m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd8f4b67b197496fb19f5ed1caa78c40 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14559fbaa33d443ea17ec08e0107d4fb : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c809ea830a9c480c95a12280b3950f29 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/562909d0abcd41c3a3ae6b7ce375c962 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728957bf5c05467c90d49c1acba973cd : SUCCESS in 5m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5daa7a52bf8497c95501e244a48a5ce : SUCCESS in 38m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ff1cf34be2f4fef9318abf502e68129 : SUCCESS in 28m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b9dd5fb138a4fa180791b7cb0ae98fc : SUCCESS in 58m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afdea543264641008b715961bef8294d : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae8870c0564d44e89a63a82183e9ad69 : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ecaadd2be0074aed9dc35447c1da0b60 : SUCCESS in 1h 06m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/7df084932e6c4399a46c3847d0962eb3 : SUCCESS in 1h 42m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c52dc1930ce5470087b44491f7b24432 : SUCCESS in 2h 23m 22s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fa6972a260a24e8093b581428eb9bc03 : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4e8df1187a404471bd5d0b4a10aee2a9 : SUCCESS in 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"eab91cf05b3ce8731bde5850fa23de563eb29204","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 18:24:42.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ac234e6f23a0b0b1d984385ee29ba0721729765a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 18:55:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/11/openstack-check/tripleo-build-containers-rhel-8-master/f61d40d/ : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":11},{"id":"abea1f73e3d8e675c7b5639ceeb4c972e46a157a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 20:52:49.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ea89bf19a1b94c4f8c330f7e4d5e21b0 : SUCCESS in 31m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/25a0355888b54243b28637f6a49d187a : SUCCESS in 39m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/da767c10b74f44e9b1d4a2071e91abc4 : SUCCESS in 1h 31m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b3a5c334188747b4b989434a05e893f5 : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cab3b59c406e4787a5f8adcfb77f5018 : SUCCESS in 1h 38m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f09aadfd95384e878f19d1b63195f7a5 : FAILURE in 58m 20s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/e180e34cacf543cab52c37cd103d73b3 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/7917a942454947e38a7071cde111c975 : SUCCESS in 1h 56m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/431c7c8c12a84371beeb11cd952b9d52 : SUCCESS in 2h 22m 22s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/239615cb72b94e52a96f1da44181c61b : SUCCESS in 1h 53m 53s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/bfa2d7ba72124a93a4673a64d81707a4 : FAILURE in 4m 42s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a5adf6f0b86b46e7a19b72c7c1cfac48 : SUCCESS in 1h 35m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ed43faed30d43d2bf9e3cdf57a58fbe : SUCCESS in 23m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/034ce3309ea4424091c928851f1f2e76 : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f9d66da7f314792b1b7340df30d2644 : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2238e7a6e541d79f8f9194963609e5 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91198ac16dc46e2880324afb79768b1 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a40c2633cac4c5b8e2d61b5a203aeb9 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/482aa97d1f7d42bdbc6d4876b46140a4 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39024fab12cc4aa794c8199609bb1e4c : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb3e43245cd946058e60ebfe6e0ba89b : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8bc94c42c8f4383b28569918f70c4a6 : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8ff44bb4dcf44fa9dc6c1953e49ab37 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85822c8441194ce48007d5c1ebea02a0 : SUCCESS in 53m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0867905b52c34e42b158f4e227355c5e : SUCCESS in 1h 00m 13s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/397a04a29ec14ec7815a9d15ec560f1b : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d8e1b20169fb49198436126e59827b74 : SUCCESS in 1h 17m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a569442dabda4833bc8d9467b1d10a30 : SUCCESS in 1h 38m 01s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7a45ff7df78249cfa99161635b7d1be5 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9299fc9fe48d44879f1d7e046a52a1f1 : SUCCESS in 1h 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c4f4739802d5e0b27b8536b3343feadb5122d3ed","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-28 22:32:56.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"253c6ab9bdb64a30cd4106f966b965600054a4e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 23:07:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/11/openstack-check/tripleo-build-containers-rhel-8-master/4add2be/ : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":11},{"id":"c89f613a21227c50fe333082aea395376d92a642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 01:00:07.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6cdcd1141b0945c394ee5ddbc86c4308 : SUCCESS in 40m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5c6c07d41ab44aada4fcf3e5de2a3265 : SUCCESS in 43m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4a76ac4fc6ad46f6a80826b8cf568481 : SUCCESS in 1h 23m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1bae58ae78d54e9caa556c0973eab163 : SUCCESS in 1h 26m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d0032ef9333f438bb8929eb4e7fc5522 : SUCCESS in 1h 42m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/95b993109c5a457bb263d4f3edf228fe : FAILURE in 39m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/6b3d19b7cc174b39938b72bc901c44a6 : SUCCESS in 40m 37s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f01925a8b24d457ab808fb46e6248bbb : SUCCESS in 2h 06m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/63a6b5a1e3694bcabffb95c3aeabef6c : SUCCESS in 1h 59m 55s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4df35226598d4d2ebb0f156cc0339a2f : TIMED_OUT in 2h 02m 16s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8d4cf46d1bde459a88b62c04238144d2 : TIMED_OUT in 2h 02m 21s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/1b2f2e8820fb4b2d9967c56eb01b2099 : SUCCESS in 1h 40m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8981644fe346454da0f10f37843832f2 : SUCCESS in 21m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6d916f2a7c1a436bb3ea385f0ac2243e : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/341b76194374451e98f5180bb48de10d : SUCCESS in 1h 05m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1a490644bea49b197a85d7ad9c54098 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85e22e9ffa54ccc9ce0f996d6938f7b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f84bfcbb611c407caa02e2f341b95ce8 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecbf1bcb7af64f56b340fc6d2d822cf9 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7cfe20833f4aa7bb3983fe931839db : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f37f3c2ee8bd40fba27f40495cec32b6 : SUCCESS in 34m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2680c2b3d9c4862b6a49196aee557eb : SUCCESS in 29m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e35eda038f624a6f97436a68578dbe15 : SUCCESS in 59m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb5c5c603a1444cc9303560d69faf5b7 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7690a2e806e400984b22e3a5f6e4bb1 : SUCCESS in 1h 01m 41s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/901098e2e06846219bf95170ca053173 : SUCCESS in 1h 27m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3e2acc98c559494abcb88316d3ed0ce6 : FAILURE in 1h 21m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/187fa3d496d54fde9effff1b8fc0bd81 : SUCCESS in 1h 50m 36s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/402f6040ad5742bb8f1d8141a50e6a09 : SUCCESS in 1h 37m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ec14e0205aac4b068e428de7ef2dbb4d : SUCCESS in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e2754c253ce1f94bb355a20d22a939f17e2d29b1","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-29 08:23:23.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"06f1e0853491e13b25d36d5b36d4b1f1de2a6204","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 08:54:54.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/12/openstack-check/tripleo-build-containers-rhel-8-master/6cec526/ : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":12},{"id":"70257322d7dd45b1ac54243da1b52860d67431c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 11:12:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e20c0f527c064361ae41a507e2762518 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a0a8831960654f548845fa5a6281f0e9 : SUCCESS in 39m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/acd787d3e04241258886254a789a7fe3 : SUCCESS in 1h 35m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/55eaea1004c64612b2b55e2e67dd7c35 : SUCCESS in 1h 22m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/054c5973420b40bea9293861ec98be62 : SUCCESS in 1h 26m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7794db10937f4965aaa372908c39fde2 : SUCCESS in 45m 31s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/fdd2425ccf984dbe80c770b8fa71fd00 : SUCCESS in 49m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/3838287c9eda4f3db670e1aeb4987cfa : SUCCESS in 2h 25m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/90c82d290c014bb6adf96270079893cc : SUCCESS in 2h 23m 05s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8880b0ca0f094f82bf6ead7376d9916b : TIMED_OUT in 2h 03m 03s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7fb57f8d13544e08885c399a5a7d9995 : TIMED_OUT in 2h 02m 41s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7eb7e5eec4534addadfc61ff2088665b : SUCCESS in 1h 46m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9550c3ff9ebe418fa37900d2f678177c : SUCCESS in 23m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/05d23173b4264b5ca755412cfa149812 : SUCCESS in 29m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be24c15366134d048ec1f0073c44eb06 : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30a5c77a9a644167b1289c5c043edfb5 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d0faf9faec4ac39e7a2c7986e6231f : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5ce2e5dbb0419b9705782f6f8ae717 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/906a83ef3e9e4ed4852c864fbfc90db0 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212492c2d26d4c78a64f915f8dbe8a95 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f803cb7094064390b170c328a3b7e426 : SUCCESS in 39m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/62048e53cd9441f780e7977ac7c5d0ec : SUCCESS in 40m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3291f0f6970e4a6d8db382cbf71bd1f5 : SUCCESS in 1h 07m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af62e938f9ca418caba456903df26916 : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0cd00682d52c4c888f533a6a7e4e8f11 : SUCCESS in 1h 04m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a48c6460eb4b4c99bd4c5862e1e307f1 : SUCCESS in 1h 13m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/dc21abcd7cea48d28828b918a37fb0a9 : SUCCESS in 1h 43m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c70ccaf624634c699960876f03d6eb08 : SUCCESS in 2h 06m 15s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/729d04bf1a7946d6914b6ffacc02959f : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8870f208ee6742cd8060e732954edcfd : SUCCESS in 1h 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"730404a9d3118536d1bcd4617cfef841740466c2","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2019-11-29 11:20:30.000000000","message":"Patch Set 12: Code-Review+1\n\nBackported this into Stein and it worked for me","accounts_in_message":[],"_revision_number":12},{"id":"006cc3211e568dfce24460cbd715a20159d2e15f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 15:41:59.000000000","message":"Patch Set 12:\n\n(2 comments)\n\nfinal touches I hope","accounts_in_message":[],"_revision_number":12},{"id":"023a4ddfb2c588742945fc0750cf4f93c7a471ac","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 15:54:04.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7924f53d4067af4fcf1926eb759f128495e1ac86","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 15:59:03.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b4d638b1c0a3455eba6408e57cc68cb65af59d15","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 16:18:41.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"822d5a09cb9cddf33cc0fca33580a7cfeab1331b","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-29 16:29:10.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"76670584f38bb0b3a375c7c5081fc73d84e7abea","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-02 01:46:15.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"b3a6d40e312355bf28f71a54d614d21b6327347d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 02:20:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/13/openstack-check/tripleo-build-containers-rhel-8-master/85c0272/ : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":13},{"id":"7d17fc4d4a8977473984cc7e0282b2c44103faad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 04:14:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c3c4d3fa660746249a5fa02b7a92d5ea : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05f028b451614a658cb37759e189f783 : SUCCESS in 38m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d311988cb19a4dffbe788075187f2348 : SUCCESS in 1h 18m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8cb76eb05c4b423a9a83c610bdadfbee : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b1f548a9cd7f4de9a3a7a594e5db3240 : SUCCESS in 1h 29m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9bbd901144cf49358cf62961cda5f04f : SUCCESS in 49m 24s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d1b1089a09214147a5aa1912cab913a9 : SUCCESS in 48m 36s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/1b9dac54467d4bcc98c5e12c57c56d0b : SUCCESS in 1h 52m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/fd77ae0635d74299adcba1befeea833d : SUCCESS in 2h 23m 08s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/9ad65661b4fd42299e73520e5c416907 : SUCCESS in 1h 42m 15s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/078fc207fc9b461e816cfca1d1e92ffc : SUCCESS in 1h 52m 09s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4a01b1dc6fcf47588be157b88d380c4a : SUCCESS in 1h 23m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/be3d43bea3544b2ebdcba9a96e923d3a : SUCCESS in 27m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3940f67a8fc7449882750641726b1c3e : SUCCESS in 21m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/669a06f67de047feabeb242e322115d3 : SUCCESS in 59m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/444197b92d364302addf63ad2fc8a62e : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2c2690b75b4e4b8ea6348d0af50265 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81e8390c08d54b86b85f1154d5bf7636 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13459edbb3bc44b7b4434b71b422a613 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba15dba76c64d47b8d17cbfb8001c20 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d9595103c5324a05aff728f2de6e6bca : SUCCESS in 22m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e83ab402352474cb4db77e1975ab284 : SUCCESS in 26m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18fa6059fcb54f62afe761bb5b0b1616 : SUCCESS in 55m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/441b5e52126f415c9a5313a76e9ba04a : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e07474cf05d94480b9bc4cede58fd0c7 : SUCCESS in 1h 04m 18s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/31dfc947a3214b3ea2a72137a1c02034 : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e667d940c77f479280376da3bf5f60c6 : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f5af09a353a04c1ca3490e4657dae5b0 : SUCCESS in 1h 31m 08s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4462f9e1b3a84c7aaf0e8eaa68d766fb : FAILURE in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9bd67607e1ea4cd5bfeba848f56ad028 : SUCCESS in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9352b9945c314139e47cfef2b44e62fc5925f2c3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 07:12:12.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"0f6a05bd77435f2541dafd56fad03d6c5166f74b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 07:59:30.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1af81766fab1dfb3a9944e0eecf1175df3053cf7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 08:06:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c2607346832f37bb7baf6d6f2d0659d15914078a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 10:00:11.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"9ec3ec04d46986b779b9f3bb2ec543ee6659523f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 11:23:01.000000000","message":"Patch Set 13:\n\n\u003e (1 comment)\n\noh well...","accounts_in_message":[],"_revision_number":13},{"id":"248a3ba5f3e646f5db88195bbffb7a8c69b354cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 12:24:35.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"8ab1c26f7c95b39a9350038405c33a95b9613c1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 14:28:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"61fb947551dfaa25b8b721fa99ef2e93d7c54889","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 14:33:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"38c6b14e385e6d1b974aba524f6a9e62e44eba57","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 14:41:34.000000000","message":"Patch Set 15: Code-Review-1\n\nyou just broke ppc64le :(","accounts_in_message":[],"_revision_number":15},{"id":"48346601dd6feaf908059d2d66becc699b5ed615","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 14:44:42.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6023312b5971a40cb167c5f6c00df0168dce4948","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 14:44:53.000000000","message":"Patch Set 16:\n\n\u003e you just broke ppc64le :(\n\nGood catch!","accounts_in_message":[],"_revision_number":16},{"id":"62aa8b22200355d7104beeb2b938b4c14a4cc31a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 15:16:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/695356/16/openstack-check/tripleo-build-containers-rhel-8-master/1ee032d/ : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":16},{"id":"228ba44ac7572d34d4164eb60a1c8af0c62598ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:15:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c13e031d4f6b43949ffaeb48ecc54a51 : SUCCESS in 36m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fdd31a9c2c3041c285dacfaa8f2f2eeb : SUCCESS in 43m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/82f51acc89f44fc2ba68c94ea149b814 : SUCCESS in 1h 46m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8e3c99cae0b9404ebe726dfdf6cfa78a : SUCCESS in 1h 23m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bc95946e3d7343af8a285eb481032876 : SUCCESS in 1h 30m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/89b4be62c72e4ea38c60d8be2e82c6a0 : SUCCESS in 55m 11s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/eebdf5c4188a4573ba3330d883d72791 : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/1c1c7aa6640244b69ae48e182949552c : SUCCESS in 2h 06m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e57cd5bfba154a8cbcabb4fdb002fdba : SUCCESS in 2h 25m 37s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c96aeaecc56f4638a01505c62405b6de : TIMED_OUT in 2h 02m 24s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8a42e128eb674adf851374fe9589d8c4 : TIMED_OUT in 2h 02m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ef25d9620129443e8ad7db8423526abd : SUCCESS in 1h 23m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d6bc5ea698fe47cdad66a0c62a814a0f : SUCCESS in 25m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f8af745c7bf4f4e912f4ecf8d444bff : SUCCESS in 22m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/795f222af249407e9cdf4a90aac6f1a4 : SUCCESS in 1h 02m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08afdb11de334780894eb85c24e1c0a5 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d607ededab49b1a240009f8857a72f : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aa3cf23538b46c0b34e55a6652b73a3 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bc118e77bcb4a75ba5e9b8afa0c6b3d : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482b8b4a2a7645d9adfb098c04f26554 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aab2202792a04129915838a9a2237f37 : SUCCESS in 25m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d1077885b4540e3bff02f73f89af9d3 : SUCCESS in 35m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f7f05eac54a4e0ca366bf13f40d63cf : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/226890284dc8476eb4297f8c4464aec0 : SUCCESS in 59m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e455dc95b80b4470b0ffa4a75f5df703 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/76955d6957af4eb4917caee3bf15e197 : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/42f04250caf64233ab557cfaeaef9d2f : SUCCESS in 1h 23m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e31f98818464481abdb712d3a799a1ec : SUCCESS in 1h 35m 16s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/16d064765e49467c8075f16cadd5cf3e : SUCCESS in 1h 18m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3c28d5ab074244398a891a70ce040bad : SUCCESS in 1h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f688f56ac665da3ea9ace1eabdb79d28ea37e4b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-03 07:21:15.000000000","message":"Patch Set 16: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"7864a273e9819b21b8fe0b62f5f7f7bb48a0c0a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 09:33:02.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"d805e3c9ce0b26e66323783fa2942617a1bbaf44","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-04 14:59:03.000000000","message":"Patch Set 16:\n\nJust commented in the bug report:\n\nI think we are reading this wrong. RabbitMQ 3.7.10 supports 19.3.x to 21.x according to https://www.rabbitmq.com/which-erlang.html. So the versions we use are compatible.","accounts_in_message":[],"_revision_number":16},{"id":"f547be5d8e0f4c262a5233d3c776e565606f4b47","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-04 15:32:06.000000000","message":"Patch Set 16: Code-Review+1 Workflow-1 -Review-Priority\n\nwe may want to go straight to 3.8 because train is frozen with current (docs changed to mention erlang version compatible)","accounts_in_message":[],"_revision_number":16},{"id":"7bfdea7cffa83a26728dc89408ba4241a839ec77","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-12-05 19:32:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"5a29b3a8c93ab3e14ce1e78489ea2e83e2c58d45","revisions":{"21af5b4f308f194a84ebff344a1730ad9ed50618":{"kind":"REWORK","_number":1,"created":"2019-11-21 07:09:08.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/1"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 07:09:05.000000000","tz":180},"subject":"[rabbitmq] Fixes unsupported erlang version for centos","message":"[rabbitmq] Fixes unsupported erlang version for centos\n\nRabbitmq does not belong to the corresponding erlang version\nmentioned at the official website[1]. It should be between 20.3.x and\n21.3.x but the current version belongs to 19.3.x.\n\nThis change introduces bintray repo to fix this issue.\n\n1. https://www.rabbitmq.com/which-erlang.html\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21af5b4f308f194a84ebff344a1730ad9ed50618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21af5b4f308f194a84ebff344a1730ad9ed50618"}]},"branch":"refs/heads/master"},"f932c5fd292c0eac7cadefcc553d68b3fd34962a":{"kind":"NO_CHANGE","_number":2,"created":"2019-11-21 07:10:04.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/2"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 07:09:49.000000000","tz":180},"subject":"[rabbitmq] Fixes unsupported erlang version for centos","message":"[rabbitmq] Fixes unsupported erlang version for centos\n\nRabbitmq does not belong to the corresponding erlang version\nmentioned at the official website[1]. It should be between 20.3.x and\n21.3.x but the current version belongs to 19.3.x.\n\nThis change introduces bintray repo to fix this issue.\n\n1. https://www.rabbitmq.com/which-erlang.html\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f932c5fd292c0eac7cadefcc553d68b3fd34962a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f932c5fd292c0eac7cadefcc553d68b3fd34962a"}]},"branch":"refs/heads/master"},"659e654df56afe813bcb55d6ba8208f758090e34":{"kind":"REWORK","_number":3,"created":"2019-11-21 13:39:58.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/3"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 12:47:07.000000000","tz":180},"subject":"[rabbitmq] start using bintray repo for all distros","message":"[rabbitmq] start using bintray repo for all distros\n\nThis change introduces:\n\n1. bintray repo to all distributions\n2. updates the pinned rabbitmq version to latest 3.7.x\n3. updates the erlang version to 22.x\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/659e654df56afe813bcb55d6ba8208f758090e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/659e654df56afe813bcb55d6ba8208f758090e34"}]},"branch":"refs/heads/master"},"23579f8453c9cf459663cac3638cef0d60cfd82c":{"kind":"REWORK","_number":4,"created":"2019-11-21 19:20:45.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/4"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 19:18:03.000000000","tz":180},"subject":"[rabbitmq] start using bintray repository for all dsitributions","message":"[rabbitmq] start using bintray repository for all dsitributions\n\nThis change introduces:\n\n1. bintray repository to all dsitributions\n2. updates the pinned rabbitmq version to latest 3.7.x\n3. updates the erlang version to 22.x\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23579f8453c9cf459663cac3638cef0d60cfd82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23579f8453c9cf459663cac3638cef0d60cfd82c"}]},"branch":"refs/heads/master"},"a1d8cdf75c81738d2fcf247eed3f039a55a85fff":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-21 19:22:32.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/5"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 19:21:35.000000000","tz":180},"subject":"[rabbitmq] start using bintray repository for all distributions","message":"[rabbitmq] start using bintray repository for all distributions\n\nThis change introduces:\n\n1. bintray repository to all distributions\n2. updates the pinned rabbitmq version to latest 3.7.x\n3. updates the erlang version to 22.x\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1d8cdf75c81738d2fcf247eed3f039a55a85fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1d8cdf75c81738d2fcf247eed3f039a55a85fff"}]},"branch":"refs/heads/master"},"739e9481d8aed65c081e0c38862dd66a7e556c9b":{"kind":"REWORK","_number":6,"created":"2019-11-27 17:08:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/695356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/6"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-27 17:03:40.000000000","tz":0},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements.\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/739e9481d8aed65c081e0c38862dd66a7e556c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/739e9481d8aed65c081e0c38862dd66a7e556c9b"}]},"branch":"refs/heads/master"},"6c22ca026cb21cfab2cae38e922b28fd743bfee4":{"kind":"REWORK","_number":7,"created":"2019-11-27 17:47:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/695356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/7"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-27 17:46:48.000000000","tz":60},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements.\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c22ca026cb21cfab2cae38e922b28fd743bfee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c22ca026cb21cfab2cae38e922b28fd743bfee4"}]},"branch":"refs/heads/master"},"58aef1d792952ed194d56624aee3118791734174":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-11-27 17:49:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/695356/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/8"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-27 17:49:29.000000000","tz":0},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58aef1d792952ed194d56624aee3118791734174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58aef1d792952ed194d56624aee3118791734174"}]},"branch":"refs/heads/master"},"baf97839450d48fd716233c000466db5c192ddb6":{"kind":"REWORK","_number":9,"created":"2019-11-28 13:32:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/695356/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/9"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-28 13:31:22.000000000","tz":60},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baf97839450d48fd716233c000466db5c192ddb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baf97839450d48fd716233c000466db5c192ddb6"}]},"branch":"refs/heads/master"},"73cebdd87cce6d7b9eda7e0a61bc9afa8c8361be":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-11-28 14:52:36.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/10"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-28 14:29:38.000000000","tz":180},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/695847/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73cebdd87cce6d7b9eda7e0a61bc9afa8c8361be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73cebdd87cce6d7b9eda7e0a61bc9afa8c8361be"}]},"branch":"refs/heads/master"},"7c68b52268bdc2933de7e181413fcac51788dd1e":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-11-28 15:07:20.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/11"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-28 15:07:00.000000000","tz":180},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c68b52268bdc2933de7e181413fcac51788dd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c68b52268bdc2933de7e181413fcac51788dd1e"}]},"branch":"refs/heads/master"},"bb836d3c5e6d6e00d31de0d6dbdd4d9467b05014":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-11-29 08:23:23.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/695356/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/12"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-29 08:22:36.000000000","tz":180},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/695847/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb836d3c5e6d6e00d31de0d6dbdd4d9467b05014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb836d3c5e6d6e00d31de0d6dbdd4d9467b05014"}]},"branch":"refs/heads/master"},"ee7355d6970594636f6b2465648f223c7db68a93":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2019-12-02 01:46:15.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/56/695356/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/13"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-12-02 01:46:10.000000000","tz":0},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/696182/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7355d6970594636f6b2465648f223c7db68a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7355d6970594636f6b2465648f223c7db68a93"}]},"branch":"refs/heads/master"},"ae6b9380744cd41a79c9a8cd6ec34bc2953a0bbe":{"kind":"REWORK","_number":14,"created":"2019-12-02 14:28:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/695356/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/14"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-02 14:28:03.000000000","tz":60},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/696182/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae6b9380744cd41a79c9a8cd6ec34bc2953a0bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae6b9380744cd41a79c9a8cd6ec34bc2953a0bbe"}]},"branch":"refs/heads/master"},"f030af2af65c28e175426db00de671f1bec6e54a":{"kind":"REWORK","_number":15,"created":"2019-12-02 14:33:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/695356/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/15"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-02 14:33:03.000000000","tz":60},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/696182/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f030af2af65c28e175426db00de671f1bec6e54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f030af2af65c28e175426db00de671f1bec6e54a"}]},"branch":"refs/heads/master"},"5a29b3a8c93ab3e14ce1e78489ea2e83e2c58d45":{"kind":"REWORK","_number":16,"created":"2019-12-02 14:44:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/695356/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/695356/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/695356/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/695356/16"}}},"commit":{"parents":[{"commit":"25d336c105daacd37a2778444b1092311a1152ea","subject":"Merge \"Drop openstack-tox-py27 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25d336c105daacd37a2778444b1092311a1152ea"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-21 06:56:02.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-02 14:44:03.000000000","tz":60},"subject":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)","message":"rabbitmq: update to latest 3.7.x (+ erlang 22.x)\n\nUpdate the pinned rabbitmq version to latest 3.7.x version.\n\nUpdate the erlang version to 22.1.x to satisfy rabbitmq requirements:\n\nhttps://www.rabbitmq.com/which-erlang.html\n\nFor x86-64 architecture we use bintray repository provided by rabbitmq\nupstream to get updated Erlang.\n\nFor aarch64 Linaro builds are used. Debian/Ubuntu get 22.1.x backported\nfrom Debian \u0027testing\u0027, CentOS gets rebuild of rabbitmq upstream package.\n\nDepends-On: https://review.opendev.org/#/c/696182/\n\nChange-Id: Ic74fe3021154826c7db577ff8d7008bfef2ccd1a\nCloses-Bug: #1853167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a29b3a8c93ab3e14ce1e78489ea2e83e2c58d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a29b3a8c93ab3e14ce1e78489ea2e83e2c58d45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
