)]}'
{"id":"openstack%2Fkolla~695417","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d","project":"openstack/kolla","branch":"stable/train","topic":"bug/1852452-stable/train","hashtags":[],"change_id":"Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d","subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","status":"MERGED","created":"2019-11-21 12:39:14.000000000","updated":"2019-11-21 16:46:07.000000000","submitted":"2019-11-21 16:44:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695417-1574354668355-752aeac9","meta_rev_id":"bf270b6952b89ef1909233bcfe646808d272d4b8","_number":695417,"virtual_id_number":695417,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-21 14:32:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-21 14:29:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-21 14:32:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2019-11-21 12:39:34.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-21 12:39:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 12:39:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-11-21 14:29:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-21 14:32:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-21 16:44:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48d588e9e4d2ed81188318b38cf56c8a1067d031","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 12:39:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bffb935ef6d84ce1ef9a12554fb83cdd2b3ec7bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 12:39:23.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"0647ff7b657d9efdda1b495e948d67875a9ca66a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-21 14:29:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fceb72b8cd9d416aa3cec7f52315a834ed5c15a2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-21 14:32:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a048af58ced7045fe3852625f0ee8d37fb2cba59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 15:08:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/79040337685f4e0681ad1dc458cb9040 : SUCCESS in 46m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/23ffbaf35ca44e65b707a2c1cba83a9b : SUCCESS in 1h 00m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f6e553ae83374b97b86b48d1cde51119 : SUCCESS in 1h 27m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9e7a015fe1f04e68ab50ee7da114d457 : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b0edb18dffb446ddb918d542032e4670 : SUCCESS in 1h 23m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/87e992ee789e464ba67c3ba3125c8b42 : SUCCESS in 55m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/463dcd8e7e75458bacc6e8674cc0715f : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/3cc2432d8df344c49f1b090069ecc826 : SUCCESS in 2h 21m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e415822b1e3f4d41a517c34379e8bea0 : SUCCESS in 2h 13m 20s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b25076d4c4e348fe97824fdc9697a87c : SUCCESS in 2h 00m 19s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f154b3a1ee104b648673f302fb97f154 : FAILURE in 1h 28m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/958dbcfa4b734212829f7f1e86694015 : SUCCESS in 1h 37m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/04d6972d431d4737a6d824f354277c34 : SUCCESS in 23m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/15521ae3a6cf42698e32ddc83280f739 : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f8913587e72b4140925a191797d4ac17 : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89f8a08716f74c2a9d6e011ee9b33911 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179d4b2b205d490da2b84bb157779c13 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f18db991d0284421ac388809529fab12 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2867a3e2205e4246b8efe098925be3ad : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f25f589e2a36472d8936afa3e6f9e9c6 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6e895acaa6408b9ebf4337ad3b71cb : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab329887240d4124b5c60f3718d9f93b : SUCCESS in 23m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d6f20f9b66c4db586f31b15fdab67bc : SUCCESS in 33m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2efa26aa06844e83b6fb7951109ce572 : SUCCESS in 1h 00m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/48363ab328014dff8997ddecf477333d : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b0025620c874d79a653d8d338b8011e : SUCCESS in 1h 12m 16s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/fe4918425ec44c7b8d971454af4d085d : SUCCESS in 1h 29m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0ac26172ed79435884676bcc495270f1 : SUCCESS in 1h 30m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/82b7cb7bb2ed4412a8de25a0b6408067 : SUCCESS in 1h 39m 20s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/48b2cb946b6e46c9888a68c779815e40 : SUCCESS in 1h 20m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b7573492490f4b26b6188b9a6e687a7f : SUCCESS in 1h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"172d15d55400a3edaf6af62edc9fd9160477c586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 15:08:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dbaa413435165e825d752c4c8582fe91d5ed319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:44:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/be2ac22fe3804ee28fc6c374c62fee41 : SUCCESS in 59m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4d815920a12449259d25bbde6c1c5ff3 : SUCCESS in 1h 33m 23s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ba51580cdd334fc092ab2081f35df415 : SUCCESS in 1h 20m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2136ba6c04df464ba0fcd5ffaa098521 : SUCCESS in 58m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef62e75c603432eb828dfc9052a578e : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfee0431cfd4b9987d3d4d513497223 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bea4e89f5874cebbe8299005c644417 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4431f227075247abbe999959e56cc1e4 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce6882e2c2ba4b9abbfa4397c0325dd2 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c013a9b3b692470da55146a31fbeb0ce : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5721e42cf3fb4e9da3661aac12d28ad9 : SUCCESS in 33m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca0a367a963a4708a9031f52f45ee019 : SUCCESS in 58m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd7998e7154a499bab93268ccbef93d1 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"70c891382081a14a3a36c5ba6ec1bc173ccdc29e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:44:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d19c134b94abd754d5c38888d493b69c71b0cdcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13c37291edc429594d1823dc83c8970 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a68269d0555ea838d468194be56efae6abfb8895","revisions":{"a68269d0555ea838d468194be56efae6abfb8895":{"kind":"REWORK","_number":1,"created":"2019-11-21 12:39:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/695417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/695417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/695417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/695417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/695417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/695417/1"}}},"commit":{"parents":[{"commit":"f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-13 15:38:00.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-11-21 12:39:14.000000000","tz":0},"subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","message":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems\n\nChange-Id: Id2463692f5a633b6e50af6e7ce6e3f49612e4c5d\nCloses-bug: #1852452\n(cherry picked from commit 19ba215f8c0de9a1a0aa8cc40d96fc841ee80df1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68269d0555ea838d468194be56efae6abfb8895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68269d0555ea838d468194be56efae6abfb8895"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
