)]}'
{"id":"openstack%2Fkolla~695518","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I4a7bbfaf3c30c57432de056f701d8abc7992d142","project":"openstack/kolla","branch":"stable/train","topic":"bug/1852784-stable/train","hashtags":[],"change_id":"I4a7bbfaf3c30c57432de056f701d8abc7992d142","subject":"Designate: pool-manager has been removed","status":"MERGED","created":"2019-11-21 16:36:53.000000000","updated":"2019-11-22 04:49:52.000000000","submitted":"2019-11-22 04:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695518-1574398126973-64a7fc44","meta_rev_id":"94afa6e25bba8b5c5d9bbe2e78674b7b325b913c","_number":695518,"virtual_id_number":695518,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 04:48:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-11-22 02:01:19.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-11-21 18:37:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-11-21 18:37:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2019-11-22 02:01:19.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-21 19:16:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-11-21 18:37:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 16:36:53.000000000","updated_by":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"reviewer":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"state":"REVIEWER"},{"updated":"2019-11-21 19:16:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-22 02:01:19.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-11-22 04:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"824f5bc8d72bb7daa34ac5d7afe4650fa9f8c0c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-21 16:36:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"552be292b0c5aae565c7f724c0675a31902eda6b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-21 16:37:16.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"7089838551ae18b36df1a6f57ac6fe34bb8511e3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 17:07:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f0a7041f95cb4b529a78e681e0dceb3de2e3269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 18:14:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0be01b19b95640d9b13601194d6c90e0 : SUCCESS in 56m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a1aac180ee2849f1acc50d4ff44cd8a8 : SUCCESS in 1h 01m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c738ba38576348c2ba6166cdfec33c5e : SUCCESS in 1h 28m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/12e238e7b87147abba148cd1c0e0508e : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e966129188f7450cac481d78a8f2f79e : SUCCESS in 1h 21m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/306306655245423b9dd52d81c221a1e0 : SUCCESS in 56m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d1de94fce008416cb0f734b93f45ed62 : SUCCESS in 29m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e17bb1ba3d445aca0229e05cfb4f343 : SUCCESS in 19m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/526b487464774a6d8687f759d1bd64dc : SUCCESS in 1h 04m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3fa4ed976e4338bab8d65635b61cda : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70041c08fda64c32811f7caf8090fdf7 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed006c61d20b4fa0ac18fade92f2c24b : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51660f9437034a6b80fa91a29012cd0d : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9624ed4692444e281444f451eb160e2 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3988952477d24123b74d171f1452c309 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/132ef91668c04bd799d881a83837d4fe : SUCCESS in 24m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcd273cda3fa44fda3097b94199d73b4 : SUCCESS in 33m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/729e1dd52a9b491f9fc9b765d09b1dd7 : SUCCESS in 1h 01m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fa546690f3c4908bf5f97cdba49c0f5 : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bee134f423d49c3bb38f6ca29c23ca8 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"771444ea21cc45fc40d872d8784d2ebb9f11e3ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 18:14:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba5903a75b25e23a2b0d919fea75a5fd965f58c2","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-21 18:37:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98805b337116cb785b30a508b5135620bb17740d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-21 19:16:36.000000000","message":"Patch Set 2: Workflow+1\n\nViktor, why reset committer? It was already fine...","accounts_in_message":[],"_revision_number":2},{"id":"acd2f11ba11b67190563019e522d39f0d1e01577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 20:12:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/124c9ade50cc4de191fead8d5f822279 : SUCCESS in 44m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a660bf38bed94e75883d08f5842fdb9d : SUCCESS in 46m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ef9cb588990045b186850953dd21a4c6 : SUCCESS in 1h 25m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9faeb800d7664559aa0c0dea7c28ed65 : SUCCESS in 1h 28m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e07f790502934bf184cd78b09e758ce9 : SUCCESS in 1h 33m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/983e32a7e6364cce9e61143707f901f4 : SUCCESS in 51m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/95e6f65c9a8546e6bcdde990a4f1bcd5 : SUCCESS in 25m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/88d8bd1d8c4942c6988a5f5bf84e7974 : SUCCESS in 17m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f630a43e71046a9ac0a91a6a9a67e52 : SUCCESS in 48m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebabe6330e4745fc9fa1e88b795fc755 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee48545ff25343bebfcb27946295e941 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37f7a72fc83e44c4bac898c3eebae545 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f654a4c77094d8c84c8ac11ec4cdf81 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3ba652ed7f1419bbf24ac6d24fed750 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbc5524530445c8b1a73c81691dd608 : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7b51138dbe54f7caa3c2a1756102ae3 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9572af9dfaec4225a549b18a61b5f609 : SUCCESS in 25m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca3f8eb2e905428b9088713febf9dbc9 : FAILURE in 45m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c1f35e322db4c688f8dea57d301a393 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9edcfe0c88e43969b5374e8b23c2c19 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8488217e65c781a6e402187607ac918ded123392","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-11-22 02:01:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"918be5fd350772a9a9ea34df4f69b2ca800bf12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 03:30:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/120d8b96443a45748b75c617e70d41bf : SUCCESS in 47m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0a9da7c8d2a4486884047b1f980b5402 : SUCCESS in 53m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2aa5e7ff84f1439989a7b81bbee6eb2b : SUCCESS in 1h 21m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d3c599f659a34b5b834e266a447a53e3 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b74c7173612e42189bf68b40081c3c90 : SUCCESS in 1h 21m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07aedbdcc42644ceb9cbc9c71647a7d4 : SUCCESS in 53m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/90148942de324fe3b84f55e4ea81061d : SUCCESS in 22m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/27c1bd283ed2448899d6432e8a99d0ad : SUCCESS in 20m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/180f36f22ce64edc81f109083f860122 : FAILURE in 22m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c23b088e6f76418e9e14155ff1d2ca8a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38580fd69c274294a4d0d04122b5be6a : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6fb51fe44b74945a30715ea83b59534 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ecf668967de42cfba1384f8903b018d : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f9fe4869e14cd7b67eab96ec25d62d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416576ae46064ea1b3d5da0a5170c30c : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5adce856fe25495ea875fec59188dee4 : SUCCESS in 26m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e082d5af1b7944e4a25b8bf1ea1b58af : SUCCESS in 26m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d15c88fa701441d0b89e12543e7c10a9 : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f74539aacf2d41ca8bdab01c8bbaad96 : SUCCESS in 58m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/904c0d3f0e184ba186c5b78105ad6402 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ab7013061010d676df08a1b415e932f8df0c20fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 03:30:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f43bb933b7a9221ae7213ddae3dbb59b645e2de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 04:48:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"540b8739c78c94f84c987a6e3957172f9311858d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 04:48:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/453a7a01659543b38a5003e3a9b585d8 : SUCCESS in 44m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/aecfe5feb5514a02b4170f1e1c8d1d0a : SUCCESS in 1h 12m 36s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4a98c28c79aa4fe88ff4cd6b1ddd8748 : SUCCESS in 1h 16m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3765411b19c240e4ac948b0b5b91d0f3 : SUCCESS in 42m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10847a57d230417cae8310ce9683a2dc : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9dbffa7c59443c8e1a6df3ddb3e984 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d57e449cea848b98c7eff45ab06264b : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d1b266f0584f779dcc748b4505fa99 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/469c28694b044238bd0d45125a073225 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10eb07580c1344f69b60154bef8348f8 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c87cd9e0fd0742fc8f08c8858eeefbbd : SUCCESS in 29m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b933ecb43c7647e593033a92aba8dcff : SUCCESS in 57m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8183817c0b4449d79b7f13884c77e195 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a4c83a71319e22c892abe42cefed4721a7eff36d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 04:49:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d080b8109174410cb4093ca33dd0d02d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ede35f87f3a81eda25bb8795179134022c9c6643","revisions":{"8354412c89a27989d53c4174413c567fc1bde59d":{"kind":"REWORK","_number":1,"created":"2019-11-21 16:36:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/695518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/695518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/695518/1"}}},"commit":{"parents":[{"commit":"f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90c1ff1bf9cb479cdbfb3ccef4ed2871a6c07f1"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-15 18:41:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-21 16:36:53.000000000","tz":0},"subject":"Designate: pool-manager has been removed","message":"Designate: pool-manager has been removed\n\nChange-Id: I4a7bbfaf3c30c57432de056f701d8abc7992d142\nCloses-Bug: 1852784\n(cherry picked from commit 63bd1c9124f6480b49885d9ccea3fc6100d50420)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8354412c89a27989d53c4174413c567fc1bde59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8354412c89a27989d53c4174413c567fc1bde59d"}]},"branch":"refs/heads/stable/train"},"ede35f87f3a81eda25bb8795179134022c9c6643":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-21 18:37:44.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/18/695518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/695518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/695518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/695518/2"}}},"commit":{"parents":[{"commit":"a68269d0555ea838d468194be56efae6abfb8895","subject":"base: generate en_US.UTF-8 locale data for Debian/Ubuntu systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68269d0555ea838d468194be56efae6abfb8895"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-15 18:41:02.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-21 18:37:44.000000000","tz":0},"subject":"Designate: pool-manager has been removed","message":"Designate: pool-manager has been removed\n\nChange-Id: I4a7bbfaf3c30c57432de056f701d8abc7992d142\nCloses-Bug: 1852784\n(cherry picked from commit 63bd1c9124f6480b49885d9ccea3fc6100d50420)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede35f87f3a81eda25bb8795179134022c9c6643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede35f87f3a81eda25bb8795179134022c9c6643"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
