)]}'
{"id":"openstack%2Fkolla~695880","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6","project":"openstack/kolla","branch":"stable/train","topic":"bug/1848572-stable/train","hashtags":[],"change_id":"Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6","subject":"Neutron: apply db migration for BGP dynamic routing agent","status":"MERGED","created":"2019-11-25 10:54:00.000000000","updated":"2019-12-02 09:53:05.000000000","submitted":"2019-12-02 09:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695880-1575280295647-8414466f","meta_rev_id":"08624efd70987bf402b170e6939b72309174c04e","_number":695880,"virtual_id_number":695880,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-26 18:56:09.000000000","_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 09:51:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-12-02 08:03:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-25 11:01:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-12-02 08:03:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 11:01:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-02 08:03:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-02 09:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1544ecf5063992c0ffb7b4f2ab5a374edee918b1","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-25 10:54:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"36fa61dd6dd41f12eaa015a233b094c5f4f8bf71","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 11:01:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"161a6114d91bd5c66108694aa1540d93cf40660a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:45:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bdb7edc7cf8e4674a20b2e9898137520 : FAILURE in 27m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cbdd7af32658414d93af6df96d6e5260 : FAILURE in 27m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dbdfa93e3ff44a058f60c75901352170 : FAILURE in 21m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b873f6f5ad7541808d2239db6e2df39e : FAILURE in 10m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0e7f0082fb404f78adf718c5cd43acf7 : FAILURE in 10m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e05f588831884cd199a9f7d0f483122f : SUCCESS in 46m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/27b38df5fe754b83b692904814c15047 : FAILURE in 5m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf79d4f8ea664a7a813f285cd05d9c5c : FAILURE in 4m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7260c7691df34eefba7117f8f8b1670c : FAILURE in 8m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68f20312a7434e0cbaaad3d68270d268 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c46274fd5b7454498f544314b6cc58d : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0488858be6643e3b67863f2e6600279 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b851f099cc4987ba249167944e62fc : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01b95b3838f84d2ba0ea1eb7938b425e : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7492568d524313a0fa230d46acdd2f : SUCCESS in 7m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4d8930ea1eb472dbff4b47e30893edc : FAILURE in 5m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/363989c1d6b7498291073289218cd61e : FAILURE in 6m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c0446d1904a4e52b0da7068e169b56c : FAILURE in 10m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f7ad4b2040048a88107c08b626c4fbe : FAILURE in 10m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cad66462f9ac48c388eed9c5a8876c71 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4de0be0db42ff92f1b9f765475714afa70e97272","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-25 16:29:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"676b09af388d750a4097105b70704dd1b2396b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:05:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d70402167a0d46c693203e484a1169f7 : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8dab3522eb874c39bf2a22896834141c : SUCCESS in 50m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc93e3646ff744839192b26ffc589713 : SUCCESS in 1h 18m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/644d1f76ce6c4c73ae0525632831b9b0 : FAILURE in 1h 01m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aaf6e1bb7a2d47dca7fb89e37028e6f0 : SUCCESS in 1h 32m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78e19ccb21a74b2483789096b880152f : SUCCESS in 49m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8560780c881d4205b28fc02663fc2ae4 : SUCCESS in 37m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e9cb961d753b4cc6a49bbef1611276fc : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0092c9c3fbff47c791bfa8b64b827c66 : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f792ebe567d7459eb5f139f0a679a32f : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0601a5640c4bc6951df63a83997dae : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f49b49c3b2aa403588a2119efeb3887c : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a07678159fb44fe8acdeca19b51cfba9 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ff211e3d03f47328e7a7a2adb5f3ab2 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7c6badf527438d8e92ab039b9b632c : SUCCESS in 7m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d426a0467264ca0b6feff0215fbdf5a : SUCCESS in 51m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e50a5e98354a48d8b75fd7a34ece71ba : SUCCESS in 1h 05m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44bd7e2f8dcc45eea2b803b0a01181ee : SUCCESS in 1h 15m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1c15734963a44f7a68d2e234299dfad : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf592ee729d249ccab89920850056480 : FAILURE in 25m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0f4ae8a8ba2e49828d18bd5ea4267b2c3f36f733","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-26 12:43:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"86194d166e92bfb3d3d8869c5756dd4165a31269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 14:29:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/29d33e028d0f4f07b844f276d0618b94 : SUCCESS in 47m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3e88d14db41a4c7f890195962037ef72 : SUCCESS in 1h 03m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ba5dec7923d7484e82ae056d6ed1cf92 : SUCCESS in 1h 33m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7632f1b593624731a46b242b6034a64b : SUCCESS in 1h 25m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/72e63d78465b40cd9ff77f372624af67 : SUCCESS in 1h 40m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/49e0b7c0ed0241b69acc52b45786f1f2 : SUCCESS in 51m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0d97086126c6438f8fc2935cbcc846f4 : SUCCESS in 28m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0c07ddc25964998a4a2132d8c2dbead : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36cfee41cf2c41878b7e3488652b5feb : FAILURE in 23m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07b2044c33e2473491be1e92e4d0ed88 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eeab069d3149248d78b19fb7b0546e : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1eddbdd65de43f88ac80e99ed52133e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9908003d3c3c44aa8536bc9484f21296 : FAILURE in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/071cb92b52694c598d0c9598b393f260 : FAILURE in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8ed8229dd54cf6ad0b7b3d348bd787 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/115a1fdf03ad4fa587fb1845458414dc : SUCCESS in 29m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcddacce824348138239a2ee6096145f : FAILURE in 4m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d71970792a343878a7a9f2c3f6089c4 : FAILURE in 6m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2852a5cc20c54aaab1dc43b46e961107 : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de5081bffaed409498ef68423e91929a : FAILURE in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"acca9626fdae7ab6f57f23f899ab67e2158c7e58","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-26 18:56:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfa1fcf2928ab60709fefd464f20524e736511b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 20:36:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/af43151357064a35a7c10595fce1b620 : SUCCESS in 44m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4999f0360c6a4b33b67dbc2f8d0ecd39 : SUCCESS in 59m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d64bfc6502514c09b1bd0a5ea8063a90 : SUCCESS in 1h 19m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b9111be4b045435fb0bb98be6a4eafa9 : SUCCESS in 1h 18m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5cc4e8c0711d4ca0a1cd1027f7d97323 : SUCCESS in 1h 33m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b0128c32b2e74686b63eb21812c14bc6 : SUCCESS in 56m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f181b48183e84a8c9117bdb2a6e2bf61 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce3a95beacc54a588c2cb8ac5cf1ca2f : SUCCESS in 26m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54cdc5d7d4c3448db8eefa11e311942c : SUCCESS in 55m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/830cf71a43164b6fb7a68247d12e807d : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb4ee619a8a4c69b4d4c003fec00a38 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3405b15da8f8479aab27cdf4e420f308 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79a6444732f947a393884ca4f3289bf2 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/300b83ae60bd4d4dac1a68e672b8737f : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9342688ccfd74633aa8c800b8200e587 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d4dedf6ca5d41c5b44eb77bbdfbc8d8 : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3dd3774d4e86402eaa2bd0dfbd6911ae : SUCCESS in 36m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a29f652a82d849cab1b23ec91f51438e : SUCCESS in 1h 24m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79a6487f523c4f2bbede4ed1de967ef8 : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7b64c4fcffec4bcf84a3d2710498cb99 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8053e002fd3b810ac4ca0010f80ee00b40110aca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 08:03:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f32316f6e61e308705e59104d592912d8ff1f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:04:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d776fcaeddea9be3aacf43a1209529ed667184d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:51:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a2260f33eb74d4a3177eb544c85409d752b8d15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:51:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/83b64592b3b94251b7ea5371fdeb46ae : SUCCESS in 1h 06m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f9800402ba644754b2a52ea7b674f18f : SUCCESS in 1h 22m 08s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/899434ff6ad147cd95ce3150433c30ad : SUCCESS in 1h 22m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e24199e07c4f407a9375244f85e1a170 : SUCCESS in 48m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bfe424ccd284ae1a55b81ef6a545b7c : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f29843ae18841d5be9a90b7b17461fd : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0e1e86f0513447dbed092140d22eb96 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1de142fec77749f6b90e5e1320cfc7e9 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1236dd1f6e4f471a920ac5518abb0b31 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e6b717f8e94f1a951a28dfb6d3a900 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30be5328b9d4476c87c058a73e171004 : SUCCESS in 25m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e38c5d3ef4b84ea48bd62714b8f33213 : SUCCESS in 58m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/109a279658424ef69f24086be02735e1 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":1},{"id":"50e0418078a91459d8271c6b5d1a6520bf3bf625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:53:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40893733e04d4180aac2197ff1e336c6 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b421830c763317d33ff1454701ac7d68bb4321dd","revisions":{"b421830c763317d33ff1454701ac7d68bb4321dd":{"kind":"REWORK","_number":1,"created":"2019-11-25 10:54:00.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/80/695880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/695880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/695880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/695880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/695880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/695880/1"}}},"commit":{"parents":[{"commit":"ede35f87f3a81eda25bb8795179134022c9c6643","subject":"Designate: pool-manager has been removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede35f87f3a81eda25bb8795179134022c9c6643"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-17 13:18:14.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-25 10:54:00.000000000","tz":0},"subject":"Neutron: apply db migration for BGP dynamic routing agent","message":"Neutron: apply db migration for BGP dynamic routing agent\n\nChange-Id: Id84d5a1b3c70c0260b85ed7c24aa7f1e007c12d6\nCloses-Bug: 1848572\n(cherry picked from commit 89372f0c5f312c222fe769a2aab8d1fce445dd84)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b421830c763317d33ff1454701ac7d68bb4321dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b421830c763317d33ff1454701ac7d68bb4321dd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
