)]}'
{"id":"openstack%2Fkolla~695927","triplet_id":"openstack%2Fkolla~master~I8484ffc8993b28fca25885d3f712ac5c0a8939b8","project":"openstack/kolla","branch":"master","topic":"bp/drop-py2-support","hashtags":[],"change_id":"I8484ffc8993b28fca25885d3f712ac5c0a8939b8","subject":"Switch to python3 in bindep.txt","status":"MERGED","created":"2019-11-25 15:20:43.000000000","updated":"2019-12-02 09:58:44.000000000","submitted":"2019-12-02 09:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695927-1575280612638-b0c15fbb","meta_rev_id":"32321c8d513c718c47dd984158f18994cabc0113","_number":695927,"virtual_id_number":695927,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 09:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-28 19:03:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-11-28 18:26:24.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-02 08:04:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-12-02 08:04:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 13:44:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-28 19:03:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-02 08:04:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-02 09:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eb2e9d47e382b28fdb615799106e35ab4387c6e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 15:20:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6301acf434315629053befdfe67a1a12750a5426","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 16:32:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/27/695927/1/openstack-check/tripleo-build-containers-rhel-8-master/693b1cd/ : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b6741f8478ca8ce51aec626300efb72b9aac71c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:07:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/33d41c4d614949cf972a7e48fbe5306c : SUCCESS in 40m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/02b1c47a016f4aa8b4a0e4b87ad8d00e : SUCCESS in 53m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3728a364ab07438cbcf7251a0e370424 : SUCCESS in 1h 18m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f4c6c574988543c4b0822fc7ca1ba063 : SUCCESS in 1h 16m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3fb0ec480561440b858215561a3b283e : SUCCESS in 1h 41m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/addd3f30f29a4399bc4ef3eaa3fd9079 : SUCCESS in 24m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/398e88b9812840729eef7c2285c861fb : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/caf0fad994de43689384dd440532b612 : SUCCESS in 52m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/738de58a80aa4c7aa8fa0dccbb41c485 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327e3cfafe6347eba967b6d3654f3df0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd6df4e74f54f1a95991ef9cb7517a8 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f215dd7e510a412da586f769203131e8 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3e11b2f4884db599ea02c663dc7b66 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f53c35e73224ec4b0704ac0007b4fd4 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba33eaf119434a2b99749fe7dd0367e1 : SUCCESS in 27m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c879c65403c24500970737e6857bc04b : SUCCESS in 34m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d87623a6eca047b285b2ed90bee24d99 : SUCCESS in 56m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/102dbd4f3ef5464dadac2dcd5207a49f : SUCCESS in 1h 03m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2fbfa9646b140b4bc1faeae94594f56 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"501ac33f586660195f3bb8ce203a77eecb4f42ab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:44:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96aec9035b7fd3a55417318cf860247eb18b9f44","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:44:57.000000000","message":"Removed Workflow+1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"9c92bb328147befbc9095897fdc5179a0f6dd9cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:45:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d0b38626dd6d8277e488e348fe984f23439851d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:48:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1274afa3a9834860b61c58a266f763c4 : FAILURE in 44m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8212b2e85a104daf9f06a74400b89a0f : FAILURE in 38m 20s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/42d5d6e66fcf4dc492c110bc5fa2292c : FAILURE in 38m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e04999baac6d460f96c890202c118b21 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e229ad143d4bed98d0b945c0d023af : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03c03ab6456245a48684a03bb58b33b3 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39643007e1ae466bacaf8e873d2efc38 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce84a02d60684f068d0b62f1f6af9882 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1243515031094f4e9c9a20c6d52478fa : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc8bdf5248a24f6cac6879686c2e3aea : SUCCESS in 39m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a75aeda4459f4fef98265e3e3522dd48 : SUCCESS in 1h 00m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6e9991f36e74969912e8ae82ceac084 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b1bb15def62ad40bdd728cfe30f3328dc6d6426b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 18:26:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5b11ea6ad6d40b343091067717a2a832b5cef467","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 19:03:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/27/695927/2/openstack-check/tripleo-build-containers-rhel-8-master/bc2078d/ : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":2},{"id":"82160e8488881aba3f360d57b0d253269779fdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 20:20:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c4897de903f5428bbd2ad2a354ea2329 : SUCCESS in 52m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5783c17611ac40918e91aab698e5e8c4 : SUCCESS in 55m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9553b8f8164349f29ebd46ed3356174c : SUCCESS in 1h 23m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2f82b6f4feac44f995f6bcb960c4a23a : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63d4e8b5ee594586867e4cbc1d2a7271 : SUCCESS in 1h 23m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0216aa5b35ed456ca2ec1fe9b2431510 : SUCCESS in 37m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0064209177e74c8ca098d43776ff436c : SUCCESS in 29m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ce46843bc3fd42149975cefb2c617816 : SUCCESS in 59m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32e4d0045d57468ba9e63eeb6a1a1b10 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f857c480a5748d68db3d717760e3ec0 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7128f92c56c642fcb41c8fc3a293c742 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8e7b141ffd14e39857de79ee28c9276 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b74e736a6884fa783c9d258ce9e89b6 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80be525ba9a143b6a70292c8582b64ae : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c345a9bedfb4da3b982bd5952f18f65 : SUCCESS in 39m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14a6b31147364f639b222144037b93f2 : SUCCESS in 36m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3180832cdf743ed8efab333b0ab6eaa : SUCCESS in 1h 19m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8df676030f3b411e8384dc3a791c6f73 : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d8a0f949cf0a4642873907aebe347c09 : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e9343e169a605a02355229f47056bf309d018145","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 08:04:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nmissed indeed","accounts_in_message":[],"_revision_number":2},{"id":"2c65ddf8552a28900ad84fd5ad4db4b3c31bf70d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:05:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64feeb0c272759c0ca0bcfe914531e40b32b2428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:56:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3d8726fb86d14905b98dd605ace1f9fd : SUCCESS in 50m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ed486e077c07444d95156fd7613672be : SUCCESS in 1h 26m 36s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/75989a7d293047a3883796a736e02dcd : SUCCESS in 1h 45m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93574a1d9b284d3a8dde5f63ee3f2f9c : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03649c16864a482fa6e191f18a8fe7bb : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf8d942a2c5f4bbdb6bca8d67dd66bfb : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47c7b4ef915e4563acda6c476fde1c45 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530fafc2532a4458aaa761ff4f3b49d8 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898a73d1012f49febbc118a2027558c0 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38a7d3ce0c0b4cfda423bb3db50a63e4 : SUCCESS in 26m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a51466f5385341f6a12ad9d6225da5aa : SUCCESS in 55m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/433cbc682c624168979031ab27fc3bad : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":2},{"id":"30f0deac5f70f5c5a16e5382f633f49d1123affd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:56:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c823b355ec946fb2fd004b58b54a0791fe707c25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:58:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835513bc883b44738436b2bad84bca83 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7ba2c5dc6d44e2bf131beb6fa1e66f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f73881a7530ee5bbdf529d69ac01b440d5b69ae","revisions":{"f3530ab05e904119aa699f2f6a6bd1d83c4ba445":{"kind":"REWORK","_number":1,"created":"2019-11-25 15:20:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/695927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/695927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/695927/1"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 15:17:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 15:18:41.000000000","tz":0},"subject":"Switch to python3 in bindep.txt","message":"Switch to python3 in bindep.txt\n\nWe\u0027ve already dropped py2 support and switched to py3. This was missed.\n\nChange-Id: I8484ffc8993b28fca25885d3f712ac5c0a8939b8\nPartially-Implements: blueprint drop-py2-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3530ab05e904119aa699f2f6a6bd1d83c4ba445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3530ab05e904119aa699f2f6a6bd1d83c4ba445"}]},"branch":"refs/heads/master"},"8f73881a7530ee5bbdf529d69ac01b440d5b69ae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-28 18:26:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/695927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/695927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/695927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/695927/2"}}},"commit":{"parents":[{"commit":"666531893cef0bf6fadae43c05d3906998cf8c6e","subject":"Merge \"horizon: do not build neutron-vpnaas-dashboard for CentOS 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/666531893cef0bf6fadae43c05d3906998cf8c6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 15:17:37.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-28 18:26:24.000000000","tz":0},"subject":"Switch to python3 in bindep.txt","message":"Switch to python3 in bindep.txt\n\nWe\u0027ve already dropped py2 support and switched to py3. This was missed.\n\nChange-Id: I8484ffc8993b28fca25885d3f712ac5c0a8939b8\nPartially-Implements: blueprint drop-py2-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f73881a7530ee5bbdf529d69ac01b440d5b69ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f73881a7530ee5bbdf529d69ac01b440d5b69ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
