)]}'
{"id":"openstack%2Fkolla~695931","triplet_id":"openstack%2Fkolla~master~I73b1964b7b22f1a2d5637209b86559b79602131b","project":"openstack/kolla","branch":"master","topic":"I73b1964b7b22f1a2d5637209b86559b79602131b","hashtags":[],"change_id":"I73b1964b7b22f1a2d5637209b86559b79602131b","subject":"Vitrage: build images for Debian","status":"MERGED","created":"2019-11-25 15:31:05.000000000","updated":"2019-12-06 21:07:34.000000000","submitted":"2019-12-06 21:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695931-1575666315195-73dee3f6","meta_rev_id":"d119d9b19273591dc0549910d25c89b2a02c7213","_number":695931,"virtual_id_number":695931,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-12-06 10:23:38.000000000","_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 21:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-06 18:15:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-12-06 10:25:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-04 14:44:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-12-06 10:25:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 08:06:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-03 19:41:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-06 10:25:50.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-06 18:15:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-06 21:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ca02c8ab254f5d48dce483f890974e71e3744f8","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-25 15:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67652a49d900092905a499e9c97c9b04f0831e26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 16:40:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/1/openstack-check/tripleo-build-containers-rhel-8-master/dabc62b/ : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"35ca095c3fe86866e1458cc8b295c1f011cfd309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a148f6716c3143d0a7fe002e1cb2902e : SUCCESS in 51m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/270673f8c9fe4cf9af99e446e1d4907f : SUCCESS in 49m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d8ef3b1c5c074fe5bf0d5a0d6b4ef865 : SUCCESS in 1h 33m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f56dafa345ca4e9eb1341316dc42aec7 : SUCCESS in 1h 21m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a0fd8ff096b046d3a65565e9ca20f412 : SUCCESS in 1h 31m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e0e208381b224f72bb12100577468742 : SUCCESS in 56m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fde5e7fbe4914ee89442954886216350 : SUCCESS in 29m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/20f2c2a4b61f4a9ba3e426b4239d7e16 : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/453eab9329104e8fae5c87a2dab5109f : SUCCESS in 58m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/451e46fb2f2d4cae8124a2132e8cea6c : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1203c4c2094c4bc099c67de53f395c82 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b509600fdfb43f6baae48b60421460b : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be21de87e60946089df4cc0f2e8af72d : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b727310752946d88d8f80d54e0bb63b : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c331c3a300d1443894330948ebc2b338 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/040495f23bc840ea83602f07da644e07 : SUCCESS in 28m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/976bea01731844fabcae0dcf1e30c090 : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/856f85cfe88d42e8bd605b4f6456833c : SUCCESS in 53m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0090ccad6194587a3cc39fef59c8b0b : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":1},{"id":"384551da8e7a8085a61dbaf98e30185fd5e78684","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:44:24.000000000","message":"Patch Set 1: Code-Review-1\n\nyou forgot to remove vitrage-base from UNBUILDABLE_IMAGES","accounts_in_message":[],"_revision_number":1},{"id":"ce02d56e85347dcd64f2ff28397755a6ba9f39f8","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-28 15:09:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e7368c0bd373818856567565b0e1585776181c1","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-28 15:09:40.000000000","message":"Patch Set 2:\n\n\u003e you forgot to remove vitrage-base from UNBUILDABLE_IMAGES\n\nfixed","accounts_in_message":[],"_revision_number":2},{"id":"0ac2e33c2361e1faefe93fa5b7d1d15bd83ba593","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 15:23:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57dd31654e7943c0df9194ad97fa6bcc3765dc45","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 15:42:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/2/openstack-check/tripleo-build-containers-rhel-8-master/c66ce1f/ : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":2},{"id":"89e68c572608800e31c489bb71b930f384ccd4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 16:34:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/735a8f187a334d74a2b7a1163e49992f : SUCCESS in 32m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f8e86acc7f614e4f85ced6d05f50a2b0 : FAILURE in 59m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/833362e690274aa89b168ac1801474f8 : FAILURE in 32m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/23d03bcec07d4160a764095a5c860231 : SUCCESS in 1h 21m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/352f1afb8cb6455e913b69349c3bd588 : FAILURE in 32m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aaea04c15d63456bb4592b9e702de818 : SUCCESS in 52m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/040515d64a0a4b6c8937d45f9a7b0c1f : SUCCESS in 23m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6906b97a34dc47a4ab200b1cdff9cd4a : SUCCESS in 20m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ee61a38cee549319cd6c706a57fb041 : SUCCESS in 1h 01m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f119bd34cf34bacb006a2b2886810b8 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1c497199194290a589425d99d16e6f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2afcdbaa8992422aa139321c6c27045f : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4126da39a690461085a65c735d7bc299 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47ad9d17a10454eaad1c617af51be78 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6422aae6734945518e3df299066f4d98 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea8fd39b53974b4f8743ac15584313a8 : SUCCESS in 25m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a588f93baf34daa8393f36a686be870 : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6460ad07c3c149a4af97ffc91e972ba5 : SUCCESS in 54m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9a9c8f5c061495685847069c8a38bb7 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b8a5a74525c8c8e7226fb33c09aaf77856494de0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 18:25:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"965a26be572b8c63b2424eb85a615fbe8ee048ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 18:57:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/3/openstack-check/tripleo-build-containers-rhel-8-master/f4b2f23/ : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c6eb6ef1ac8a6243c2da565b15a64e27602f9c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 21:29:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cd395b878b0a405b81ea49f5503ab90d : SUCCESS in 30m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0784921c32f94c46b7c5bdc1d6bb40ca : SUCCESS in 36m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/95c1915603404cc3a17b291547737366 : SUCCESS in 1h 44m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2c42c91d7cd94f27adf82bbef499ea0b : SUCCESS in 1h 26m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/866f5aacabdc44d98d5eb241f500bceb : SUCCESS in 1h 57m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c3d73767aa6745bc99206b94589084fc : SUCCESS in 1h 01m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/494c32290fae46189b0c8a65ccb50ff0 : TIMED_OUT in 3h 01m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2fc35e8cc4d0434685d046824e21953f : SUCCESS in 37m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c21929d33a084e1791faf28cc0f2afd0 : SUCCESS in 1h 13m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a821b3845263461e9e3447bc9aeec75b : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704803cc67b140dbbaa052cc0d1871e9 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2d80df82ef47d2b6e43cd13a031cee : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a72eea1ed0414482bf18c1e13672487e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29aead38f7d4b69b4761f875fd2873a : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef84345af9064505b126326bc99eca3b : SUCCESS in 26m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/156e4e44d56f4aaf9fe610c4e63a59a9 : SUCCESS in 27m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c48699cd00942f4b7f88e672c066314 : SUCCESS in 57m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9fa4b5a11a5a49f9b2c5cfb1cb831137 : SUCCESS in 55m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/626ff1e399144ec4aab94d0408aa4c89 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":3},{"id":"afaab0b5f678deaad7541a7d590e67585d7ffd7b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 08:06:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0fea33b085b44b6e97f76e9f906993227a92ef91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:07:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d1ffa6a2db0a1f22856159e0928f338f8ecab2d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:59:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/21b59ab5bcff4b85a0bf0027ba0844c5 : SUCCESS in 51m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1824ac4893924005b86afb9103b6a369 : SUCCESS in 1h 26m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/50297289f0674bc7b368ac22dcbb00a3 : SUCCESS in 1h 40m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/49ffaabea66546c7b3097b4bd985b8c4 : SUCCESS in 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/202f9f2a3ead43d8a32bbac7e526e8b7 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f8129979e5b4579a2f439f6fbf2a908 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c4b957983a402788f069f211369700 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de2d52a04fba4dc5b7aed5620fa8e45a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8836a6f069240e68d2b9067c47bf863 : SUCCESS in 7m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f89b5f7d9d04c0080b36bd88db2bcb3 : SUCCESS in 36m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdf6daeed07f49c081ff0d81b5bfc554 : SUCCESS in 1h 12m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c073eb9d24e421d8e667c3b58e65061 : FAILURE in 1h 27m 24s","accounts_in_message":[],"_revision_number":3},{"id":"3debf897d844535692b280ab8410f775720e8583","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-02 15:48:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2d030a839c4118f815b6119fdce4c63be577ac03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 16:25:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/3/openstack-check/tripleo-build-containers-rhel-8-master/7eb002a/ : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":3},{"id":"1ec114734d7d3ea2f4c7c1cc75cfd421a884bea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:33:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f18d3cffa4ce473bb91b922166855926 : SUCCESS in 34m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f7e0e7b7129646d5ada9aae6ed595698 : SUCCESS in 37m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4d6e573e54364e0eb67551f726985496 : SUCCESS in 1h 35m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fcd6bde1bf9d4ec59c9d4eb88a7d55b9 : SUCCESS in 1h 20m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/77fb71ac75c94ddfac39c6dcac807a66 : SUCCESS in 1h 40m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/97dbc512fd83479993d82ce0bc052ff5 : SUCCESS in 50m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7680a2c5e74c4a87809999bd99ba9f17 : SUCCESS in 25m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/201d2416a51e4a108be6b1bf8cd15bc9 : SUCCESS in 22m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b28018693f7346ccb7532cde3eff4a7a : SUCCESS in 1h 09m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43fe1a70fe544544a724d9beeabd913b : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4148c9e209104f84abc77dca0aad17ae : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d23d6aab659a4a6bb5795a619a713c54 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5312f0851dd472cbae620799223287f : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dae7c7264cc437ca44924c77db28940 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52ebd05f6cec4b5e8947195562b36bc3 : SUCCESS in 21m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ee7fc29532b4718b8c4ea159ad343a0 : SUCCESS in 26m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/721ea5f348b94dfb834c4294d991239a : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bfa0615ed05f48eb9776c3c4dca2b2b7 : SUCCESS in 59m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6cdfe2a6a73b4b6db36e3fd721ee20ad : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":3},{"id":"dde93033a23e39c11c639390154c2421dfb40465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:34:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b1b35803d940d40e3601398419901ecadc6c74d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:45:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd8fb0f8aa8547eab3f2ae4d5412e17d : SUCCESS in 49m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/35c71b2080864b8981a701115b7a63d8 : TIMED_OUT in 2h 01m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/93d7bbe747174ca1b929e1570343ea52 : SUCCESS in 1h 41m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eab2cf507a674e9ca07740481ad107e5 : SUCCESS in 49m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5c2539567204d53964bcad6c8fd501f : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7faea5cd674e089e90d6761d3f638a : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84f7fc3ae2894daaa0192382d401cf80 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2af8728cafe422c9e059f29279d0543 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0149666a4f47cc8d1980f13a9a3b52 : SUCCESS in 7m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/364c9305681047fdb5f4a547c984cee9 : SUCCESS in 39m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a17d6100829497ba0469aa48f6cd287 : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4534f2a2db234e99852692afda5a9216 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":3},{"id":"68d038f1f66bb6a467af73decf6f144dd62fc6b9","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-03 16:04:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"abff7437770bc4ed8e8f5d84154a85012f29cbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 16:05:04.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"5f8e71b5b5f67035647a4d3614c09adcdb03fbd5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 16:05:08.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"0b3b32edddeed550c26744a638a3d8db5668feae","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-03 18:54:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dd4d82706ec653c4432ba170d8b68a0afdc17f36","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 19:28:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/4/openstack-check/tripleo-build-containers-rhel-8-master/32f17ff/ : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":4},{"id":"38ea33849c869d55fa3ff0d59e56ba0b39c09f49","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-03 19:41:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(rebased)","accounts_in_message":[],"_revision_number":4},{"id":"4abdadd5342f1abeb20f0a845320384145539095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 20:32:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e0bc367eac29472a90001abf3f7b69f2 : SUCCESS in 30m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/53588eba3c894b64b8632aec14ab6593 : SUCCESS in 40m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/746fafd2ff824778a05e937e574c76bb : SUCCESS in 1h 17m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/90deb7d57eb54743b3654783bfe4a085 : SUCCESS in 1h 25m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b0ac011c2b6049e7bfc4be69b8eef34b : SUCCESS in 1h 34m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c0feca08edd947cb84acde933ad6db0d : SUCCESS in 52m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/11a7bcf365064563ade7c8b658697a83 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/daa0387e182849b7835b5b6a5f573b54 : SUCCESS in 26m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d6a48410e800462b9365bf80a4f8b104 : SUCCESS in 58m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84a7fec3f28644b499c5bed18a0d7a6d : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6068641d9e36498fbd872008051eb0a4 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98ffc26e56cf4625ba0fcb7360c5aa85 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9976e1cc47c74cdb935e73a452e67867 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f95a59c4f8446dd9f22e949d59468de : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afdb09d8fc9746639edef7f9d15c2891 : SUCCESS in 28m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c013cb6e0b74d12ba9eed5b0a60b5db : SUCCESS in 26m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2e47f4c939944748c43c5b9c347b645 : SUCCESS in 55m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3e2c089123741c6965038a9b4d2ca1f : FAILURE in 53m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/68717935b2554952b83e7fc0827cd207 : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":4},{"id":"f66d97335af2b1b727a73e70bd914d76798c8d68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 20:32:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a269b85d18626e24cd95e1383fb4aea07c6ea79c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 22:07:04.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ea19e7d34c84487c9beb19d0d719c2de : SUCCESS in 40m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/abbea65aeec245e2af95c6de7c0d81b7 : SUCCESS in 1h 22m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5afb72addd714730a159da913c581770 : FAILURE in 10m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5ceb45169d894fd1b899a30039aaa24c : FAILURE in 15m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9db3bbfc9842989a076346704fb1c7 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0bb593667f4758a132bb1740b40a8d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c334428914814035a4a930ab0e70b2de : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/376501b7cf4c4bfba137814b5db28cdc : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e465a5f75f1443758af20b199e552ee9 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/641ec54a4d544155bd925e5f03cc9ee7 : SUCCESS in 39m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33effdd903bb4677a56a014416801940 : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/153fbbdcd0a248d98ccb38d7b7ef8c1a : SUCCESS in 1h 32m 53s","accounts_in_message":[],"_revision_number":4},{"id":"76bf9bc976e0bf82c5154d6cd9cf60e9cc185a04","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-04 14:44:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4b2f6a5a541678cc0b631599badf07c9155b22d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 15:14:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/5/openstack-check/tripleo-build-containers-rhel-8-master/bf8c7c6/ : FAILURE in 29m 31s","accounts_in_message":[],"_revision_number":5},{"id":"2b6cb88c0cd914e5caee573038ed0f93e321f170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:56:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b5a35b46264a4ab0a1319484ad0df5b2 : SUCCESS in 35m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e057c72a2117404e8702496172e91fd0 : SUCCESS in 54m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bfefcbfcb0874e0bacabcbbcadbf8bec : SUCCESS in 1h 30m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a1babdeed91242a0adc05b51b3c9c050 : SUCCESS in 1h 27m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/438a147862e24ffcba82c311478cf974 : SUCCESS in 2h 05m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/786fa2be76f64c0baf95de3ca56dc995 : SUCCESS in 1h 02m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/146eab55d12d41b49c1e6270bde45fca : SUCCESS in 32m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/91aab4c4f6b64fd3a6ff7b220e1790d7 : FAILURE in 26m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd12c914d3e0481aacd8ba49af7305a9 : SUCCESS in 59m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e31b5becf0842a8b32ff94652391ba1 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50fcd0260caa49caa97987d93ed6976b : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e22cfbf61c4484ac46e9c53739296c : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6c2bd33f176498f94df67d565a533f6 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03663a59646041bda1758c8a0d15d0fe : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27a6e26df13748f29876002e17f83f56 : FAILURE in 22m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/26e7e0738842437da8d8d3f9a49bd75d : SUCCESS in 30m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2a44445e358430097c8b607580ea819 : SUCCESS in 56m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/976dc4cb1baa419390b7d2cab71299c7 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5e841464b1c4059a084f385a52af207 : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e30c8b45a00072c09b8c353489953ceb3d1d69fe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-04 20:11:50.000000000","message":"Patch Set 5:\n\nit does not require uwsgi for binary type?","accounts_in_message":[],"_revision_number":5},{"id":"073eefbc9ec5f68f89594b064738f7afd2c93b05","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-06 10:23:38.000000000","message":"Patch Set 5:\n\n\u003e it does not require uwsgi for binary type?\n\nyes, uwsgi is not required","accounts_in_message":[],"_revision_number":5},{"id":"c2cfa786ba3bab721da4a363e9bc9a900fa51a83","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-06 10:25:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"69c4466cfc6af74f5535653865a4a004c848c53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:26:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bc7297564aba031344b83bc3170389f3a4daa102","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 12:17:30.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6b06c49f0b87427ebd4fcad381b06055 : SUCCESS in 38m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/87482d7213b64fb3bf604fa959b54eec : FAILURE in 1h 00m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/de8c400639eb4291a0f9aa4d88ef1768 : FAILURE in 1h 49m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5589fae7351248f0a8461a314d3a0c94 : SUCCESS in 59m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb67565cbb85459eaad075079669a55c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6094fba9384510bfbcd47ca1c94aa5 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d28ba16bcc8e4f84b12771a2abc8ff4d : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c56c5f4c43f24846aef32e5bee62bffa : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473176791624408981727e3e6bd10124 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52793b1a62b24b54b3f69a5e7a83dff9 : SUCCESS in 29m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c57b6a542a74217bee363a7707ba232 : SUCCESS in 1h 15m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bff1de07690c4e4ab1a10f73d63585cc : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":5},{"id":"fddb41495a86ab5cc7d97a6dd539f73fb18d8618","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-06 17:41:01.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4a0c37e789337447856573ce463dc57d2ebe9c9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-06 18:15:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/695931/5/openstack-check/tripleo-build-containers-rhel-8-master/89605a7/ : FAILURE in 31m 43s","accounts_in_message":[],"_revision_number":5},{"id":"495cdbfe3644e9f51b7fc6b6058a7966aef41ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 19:26:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e6ed67f5d7dc479d9f6d90ad91edc424 : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/47a4cb909b364f0cbe178b96987e912a : SUCCESS in 1h 11m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6ff8258fbb654646b6c35bcc97253f74 : SUCCESS in 1h 15m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/017fa6e68fdd4568943c614413da7707 : SUCCESS in 1h 34m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c1dfb86b11494c26afba96b48e7ec94d : SUCCESS in 1h 44m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2f0cf05ce5934eae9d46389dd9a833f1 : SUCCESS in 51m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb4e8c63fffc419cab8575d437ecdaf6 : SUCCESS in 23m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b88d53e806004892aea83fff842956e8 : FAILURE in 18m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad31878b90a94d95935ce8e3236e2fe7 : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23b65d3c445e4316a3993f4b6d54f744 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc630e432e244cfeb176c91cc4941161 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3917aecac638452e9cd685259aa9142d : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb521a0271e84f21b9e4f783c6031fde : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6d63bafbaf4c7a88c647889d5ccb4f : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99e9b304c80844ec8bc24afbd3565695 : FAILURE in 25m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c51a610f2eb4c9086a268ca21952e42 : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/db8ec64228794d3baceed3297a1f0653 : SUCCESS in 53m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2920b399ee6245b890cce042eaacea85 : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64c3b548982b45faac949aea19644c0b : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":5},{"id":"762e3488db135b8541c183e800d92df75099f0d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 19:26:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1816abf8cd6561ca9c91541c6d11f788a52dd40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:05:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a8cce6a40f34839ac35e0d8d333727b : SUCCESS in 34m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e0f0112a68d44980a12ad48dfddbb350 : SUCCESS in 1h 25m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4262e8480dc747499e7ea43e029d4877 : SUCCESS in 1h 22m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c9277b1a53b147f2b669dded0a2cc79c : SUCCESS in 1h 02m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f42b42543964c13b55cafb0f2d0108a : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b77c7688194ca69e15802ab029b60d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4d37db7a6e24915b88953cf4568ba8f : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1037c32172547389f98089fae45a9ea : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62afd43bbea447d9a838ab5688fa4850 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bfef7a36fb94a05b3cf78d0d87c5cdd : SUCCESS in 28m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c81663ab2ad6430f9cd306fc4456ac89 : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca690ff049c347f0863c5bb920920f1d : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":5},{"id":"2ded3f1898d7bbc0d7c72642c0d2b70cae3e88d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:05:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9af942cf7b978a4515ba3646ad6e6d74ec12e8cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:07:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8e8a8e26c04b3eae195d485d50080b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"140ee6c83948f4661c04c464bb1d0d9cf2573026","revisions":{"4a2af9267a4792e2375b8d8963a5d4567afd23fe":{"kind":"REWORK","_number":1,"created":"2019-11-25 15:31:05.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/31/695931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/695931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/695931/1"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-25 15:30:24.000000000","tz":60},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a2af9267a4792e2375b8d8963a5d4567afd23fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a2af9267a4792e2375b8d8963a5d4567afd23fe"}]},"branch":"refs/heads/master"},"5e42a70f2db93effbfe813db2d52b75ace0e8e47":{"kind":"REWORK","_number":2,"created":"2019-11-28 15:09:13.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/31/695931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/695931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/695931/2"}}},"commit":{"parents":[{"commit":"1dd003a24b874b5f3715e7c661489f6ab2718bcc","subject":"Start README.rst with a better title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd003a24b874b5f3715e7c661489f6ab2718bcc"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-28 15:08:55.000000000","tz":60},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e42a70f2db93effbfe813db2d52b75ace0e8e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e42a70f2db93effbfe813db2d52b75ace0e8e47"}]},"branch":"refs/heads/master"},"bbbf7cf99612018f3dd93c65bdbed71fc7467575":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-28 18:25:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/31/695931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/695931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/695931/3"}}},"commit":{"parents":[{"commit":"666531893cef0bf6fadae43c05d3906998cf8c6e","subject":"Merge \"horizon: do not build neutron-vpnaas-dashboard for CentOS 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/666531893cef0bf6fadae43c05d3906998cf8c6e"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-28 18:25:52.000000000","tz":0},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbbf7cf99612018f3dd93c65bdbed71fc7467575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbbf7cf99612018f3dd93c65bdbed71fc7467575"}]},"branch":"refs/heads/master"},"9ead849f4e6d322041e81d63924b2685bc1f5a67":{"kind":"REWORK","_number":4,"created":"2019-12-03 18:54:43.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/31/695931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/695931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/695931/4"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-12-03 18:54:25.000000000","tz":60},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ead849f4e6d322041e81d63924b2685bc1f5a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ead849f4e6d322041e81d63924b2685bc1f5a67"}]},"branch":"refs/heads/master"},"140ee6c83948f4661c04c464bb1d0d9cf2573026":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-04 14:44:21.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/31/695931/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/695931/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/695931/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/695931/5"}}},"commit":{"parents":[{"commit":"59c881d85c4c9154a459c8c695f10243b387f047","subject":"Merge \"Added senlin-conductor and senlin-health-manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59c881d85c4c9154a459c8c695f10243b387f047"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-12-04 14:44:06.000000000","tz":60},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/140ee6c83948f4661c04c464bb1d0d9cf2573026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/140ee6c83948f4661c04c464bb1d0d9cf2573026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
