)]}'
{"id":"openstack%2Fkolla~695937","triplet_id":"openstack%2Fkolla~master~I4e60fcd767f15b7da90b0143008b1d6d794c76f2","project":"openstack/kolla","branch":"master","topic":"I4e60fcd767f15b7da90b0143008b1d6d794c76f2","hashtags":[],"change_id":"I4e60fcd767f15b7da90b0143008b1d6d794c76f2","subject":"Horizon: install vitrage dashboard on Debian","status":"MERGED","created":"2019-11-25 15:50:50.000000000","updated":"2019-11-29 21:24:36.000000000","submitted":"2019-11-29 11:32:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695937-1575027179516-eb3ba9e4","meta_rev_id":"94268f2a7eadbf56674d88248af58241d62e3cb4","_number":695937,"virtual_id_number":695937,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-29 11:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-28 18:58:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-11-29 08:58:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-11-29 09:06:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-11-29 09:06:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 18:58:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-29 08:58:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-11-29 09:06:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-29 11:32:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fbca6339b6d885893a5f3cf7ffc243e36ca4926e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-25 15:50:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03b8bcc33e0c807eaf7bc4e8db0bef01db6fd53c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 16:46:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/37/695937/1/openstack-check/tripleo-build-containers-rhel-8-master/8e71528/ : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2141c3374abbb83e72899fb39cb96d10c384a675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:35:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/390225a9f7424cdfbf9e71c546495020 : SUCCESS in 36m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b49f8b252c024a9fa03c77f7369f9aea : SUCCESS in 1h 05m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6c739c31437e4c20bc714f0b1ae422f2 : SUCCESS in 1h 28m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3cb70b9c60984b419211790d262f6494 : SUCCESS in 1h 29m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e580725a08814781b8cf407f49f08021 : SUCCESS in 1h 40m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aab3423c3c484fb0a3b5b44fbc7d89de : SUCCESS in 44m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef73fe45a18b497cb02e84f698efe429 : SUCCESS in 22m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/878dd5055dfc4ddfbb54a2fedc42c752 : SUCCESS in 22m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1794b28e01504a909d50246486a47524 : SUCCESS in 52m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1ebfb75f184a59b73aa2f3cf1aa25d : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd2a688e0a1408294a20381d2b781e2 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07449147e02c431f85bff2dac553275f : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e89250550d6d4707883be80df4302e9c : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861d931e46be45e8b641aa8586e4dc2b : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5962253daf3644b9a315e058c8963e88 : SUCCESS in 30m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71840aa7627a46639c15b84ee2106dc3 : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a6feafc2cc349dbb4774e69f30805fc : SUCCESS in 1h 03m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a6efdf0a9104319b95ae023ab50e67d : SUCCESS in 57m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd1df456f4414e80980331d7e51ca7e3 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0afc456d47e519ab3c379d8076167030500af7a9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 13:42:47.000000000","message":"Patch Set 1:\n\nwhat about other distros?","accounts_in_message":[],"_revision_number":1},{"id":"72293071319c77eb173687de25e6b03e9acac5db","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-28 13:54:53.000000000","message":"Patch Set 1:\n\n\u003e what about other distros?\n\nThere are no vitrage packages for Bionic at this moment. In RedHat part I see \u0027openstack-vitrage-ui\u0027.","accounts_in_message":[],"_revision_number":1},{"id":"7c1e5631e644ae402740a868fcd2f558ebf67298","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-28 15:10:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a9d965856fe1cfe2a676045af39a2fdd6c99121","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 15:23:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0423aa1634f482f80f81ad7b09ab39099f6f19c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 15:47:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/37/695937/2/openstack-check/tripleo-build-containers-rhel-8-master/5b8cbd9/ : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fce86108fbb6c669ea7fc0be94507ea6ce9f3312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:07:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/054ba9b7c32a445987a0032c69dc62c5 : SUCCESS in 31m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/23740bd0a8ce4cbe9d3dae6099bb3779 : FAILURE in 39m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/088043a6272c48b79dd8c8f21e89eb21 : FAILURE in 43m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b3cc7886ce6c4d468dffb0b4f5eed1b7 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/86c1cf614b264d72b584d15eff7294a9 : FAILURE in 28m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7d441bd626cc42019fb6fe21eabe3a7a : SUCCESS in 42m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8223a1069fc04ae0b742b80560215ba4 : SUCCESS in 29m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29c25e6cab014df0b40f6ea7a30cc1a8 : SUCCESS in 20m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a2a3672f3f9c4d848b194fa0a8f0538b : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a73a49a585f432fb3e8728393dba0eb : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d3898450ea4fce888f9b7ce77b18ad : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/446ce68eab684bb58f77d2fa6ce26829 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a8ab25a73da415ca311d761c5476397 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8425609e85174fb1ad082e791f397398 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7628316a705466cb9428849393ba321 : SUCCESS in 34m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3719b60a7548425ab219b9eae54af00f : SUCCESS in 28m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/05f00c4683794bf6b48da0db50a69939 : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9a3befbc2024b2a9a9046d47ec0d45b : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/703707653c624b5da39afa35067d8196 : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":2},{"id":"92c403a102afe883a0dcb5688867258011593855","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-28 18:26:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"db0cf83cb3ba0d246029792628dd06b9af8565c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 18:58:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/37/695937/3/openstack-check/tripleo-build-containers-rhel-8-master/ec0920e/ : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8412d9885bb0c846ff84d3c26da4099c72fc72d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 19:57:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7759dfec72994484b9d27b8243cfd132 : SUCCESS in 42m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ed1a38c03b924f7999ed9460abac72b6 : SUCCESS in 42m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/414b7ea11222462a91e6aff3c0ac1594 : SUCCESS in 1h 23m 53s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/db6adcc845514e6498213d72f6690315 : SUCCESS in 1h 19m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d3e23f7c4cc942068d8e91f923d928bb : SUCCESS in 1h 26m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6a55f2e7bbcf43dbb919d379a7f5ecc2 : SUCCESS in 42m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9104eccffc984f11a3e6cd8e07bc8b02 : SUCCESS in 36m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45e7e1a840e84dab981044ac413c6830 : SUCCESS in 33m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6c275182b964ad68b028c41f2464895 : SUCCESS in 1h 04m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21674f65889f4708a16a08f6985aafe2 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595aa428ea7e49b98bd5d728fd7066d9 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4861fb6bf9e74305bb11be0fd9501c74 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8db58c4238f741ef8e6e2f9a9af2b452 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fa46730390493ea55072a4977868a2 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/373e0605b1fd4b34afea9407f387775c : SUCCESS in 29m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10bf151fb73f40069e5eb2b942331f43 : SUCCESS in 41m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3572275e3bf45909f86c08d2f42320d : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56cf7ba45c8d4b11a6ddbacaed7f2d91 : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d56a01ab9c444a1ca8e76512c74188dd : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":3},{"id":"75704add25d97a8641a0e0933663a6e1251133ec","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 08:58:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1f6e530cafc8f048e4b74d0facb8477106c94d5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 09:06:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4a10358812067daaaf7bb5065bd466bff6d29cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 09:07:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86a001d9a8ba654e7130b2f7809e1b10ec64d8e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 11:32:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fc6680ead49c484db86fb1791cbd0420 : SUCCESS in 45m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e58b832ef2524c32b8c937afe2cbbd64 : SUCCESS in 1h 33m 42s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/03662d2a88734b2090f86f16f1f6e319 : SUCCESS in 1h 42m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0983bae5209a4fc888fa284ffc637098 : SUCCESS in 57m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47647fb56df64ba3bbeaf7dabd3a1c1d : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639f85e99f634003bebbc0cecba32b25 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16680224b04d4a62932354e9a60f9cad : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6819b61c311446719e8bdd1cb28bafe9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec97d55bd61c4b86afd6905939b3f4bd : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09ad299af12646bf8fadd841d7195d07 : SUCCESS in 28m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/393e790fa0ba40a3a8477efa4bf5e8ee : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd7fe361a1174067be659caad2c79131 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":3},{"id":"312e96fab8f0d01456ac1c5fd9725b343f8fa2d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 11:32:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"07c5971d7d6b160fe17932cbf8f23dd1cfd382da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 11:34:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1863b56e7f41798439e193c1714a3e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3},{"id":"1607f7c836e4b4d7b2772ecb874cc30c917b4430","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-11-29 21:24:36.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9bd4a3a21cf17cabce845fce4860bbf14f993f02","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"683f9c9178ec541d2ab8f92883da66d8e2f1b956","revisions":{"a93517d7ef898e5ca1946162628f7a19a184827c":{"kind":"REWORK","_number":1,"created":"2019-11-25 15:50:50.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/37/695937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/695937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/695937/1"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-14 14:46:27.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-25 15:50:39.000000000","tz":60},"subject":"Horizon: install vitrage dashboard","message":"Horizon: install vitrage dashboard\n\nChange-Id: I4e60fcd767f15b7da90b0143008b1d6d794c76f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a93517d7ef898e5ca1946162628f7a19a184827c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a93517d7ef898e5ca1946162628f7a19a184827c"}]},"branch":"refs/heads/master"},"b0b326774c98d744d347e987a39c608a54ae528e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-11-28 15:10:55.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/37/695937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/695937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/695937/2"}}},"commit":{"parents":[{"commit":"1dd003a24b874b5f3715e7c661489f6ab2718bcc","subject":"Start README.rst with a better title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd003a24b874b5f3715e7c661489f6ab2718bcc"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-14 14:46:27.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-28 15:10:37.000000000","tz":60},"subject":"Horizon: install vitrage dashboard on Debian","message":"Horizon: install vitrage dashboard on Debian\n\nChange-Id: I4e60fcd767f15b7da90b0143008b1d6d794c76f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b326774c98d744d347e987a39c608a54ae528e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b326774c98d744d347e987a39c608a54ae528e"}]},"branch":"refs/heads/master"},"683f9c9178ec541d2ab8f92883da66d8e2f1b956":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-28 18:26:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/37/695937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/695937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/695937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/695937/3"}}},"commit":{"parents":[{"commit":"666531893cef0bf6fadae43c05d3906998cf8c6e","subject":"Merge \"horizon: do not build neutron-vpnaas-dashboard for CentOS 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/666531893cef0bf6fadae43c05d3906998cf8c6e"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-14 14:46:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-28 18:26:09.000000000","tz":0},"subject":"Horizon: install vitrage dashboard on Debian","message":"Horizon: install vitrage dashboard on Debian\n\nChange-Id: I4e60fcd767f15b7da90b0143008b1d6d794c76f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683f9c9178ec541d2ab8f92883da66d8e2f1b956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683f9c9178ec541d2ab8f92883da66d8e2f1b956"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
