)]}'
{"id":"openstack%2Fkolla~696710","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I86b65db4a7760c2f6d2494266389dce518df89de","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I86b65db4a7760c2f6d2494266389dce518df89de","subject":"Bump OpenStack versions for Train","status":"MERGED","created":"2019-11-29 14:46:48.000000000","updated":"2019-12-02 09:31:22.000000000","submitted":"2019-12-02 09:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696710-1575278996705-cf5fac72","meta_rev_id":"0132703a9e405d4db622d803ec42d27121992d59","_number":696710,"virtual_id_number":696710,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 09:29:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-12-02 08:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-29 14:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2019-11-29 15:42:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-29 14:57:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-12-02 08:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 14:53:09.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-11-29 14:57:49.000000000","updated_by":{"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"},"reviewer":{"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"},"state":"REVIEWER"},{"updated":"2019-11-29 15:42:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-02 08:03:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-02 09:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d928e355142585a9139ea230f2458ac2c9a505fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 14:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e620ccf1eb5eec3f2d02a8b01657c077e655e970","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-11-29 14:53:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72872bdb84dd4df15f7631ae81193aab2a333a55","author":{"_account_id":30927,"name":"Andrei Nistor","email":"andrei_nistor@smart-x.net","username":"andrein"},"date":"2019-11-29 14:57:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9240c4d5ef0476c3e29cf6adcf435bd0541e9231","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 15:42:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d229b15a85eb46eb48d8e2c9ab922e255c911d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 16:18:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/089e98b97e9e4d1788b45bd9ce499ad4 : SUCCESS in 49m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0887f5b2cc9e408a9b52558f052b32ef : SUCCESS in 1h 06m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e890f1b3c86f472aaf92baf3434e48f4 : SUCCESS in 1h 30m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9f81e83a97084b3aa6ae05cea0194b1b : SUCCESS in 1h 14m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1ed79e994dd44ac79ee112d73168d82e : SUCCESS in 1h 27m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9d3f39b9efa4451ba4a1f31e0a15f852 : SUCCESS in 42m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a568e9efb7b04f5790c2082f6f3426ff : SUCCESS in 46m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a36b760f809e4dda98d49259b1b22752 : SUCCESS in 34m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/befc6f6df23e462abce9267e237c2260 : SUCCESS in 1h 05m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f5342d1571045b99ad4ec3dcaea38ff : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ff2b5e270f4114bfc53e0bc2abc81f : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd49014295cc43cbb3835f8c958ac8e2 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4d28aa14a5451483a2f3e72a163a59 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ae8213961c747e4b99631464e1780a7 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385c704df0fe499e838629150512525f : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c1ac057858d483e946cee929d1b1367 : SUCCESS in 24m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e8d5ccf67ea4b8bb75f7f722ea13725 : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1f4eb12d1b2482eab7c56fcb6f03ebb : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0193a522b3a6496a82a6730267a001f0 : SUCCESS in 1h 00m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8342bd263c14cf6ade51efa02449de7 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a8ebf6333f77bf999ac8b2584897f5a94db9e1f0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 08:03:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"caea614dcba24665a2c99c3045a73ad8994dfe9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"889bd876b8e970403fe2c8baa7f407dee8658485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:29:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/056cb5c12d564d8fb9380f744ddce3ad : SUCCESS in 47m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bf40f044028b478b8afdb8ca0719413e : SUCCESS in 1h 22m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/19772917f5774a828519fc52503780c5 : SUCCESS in 1h 21m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/021a45701a574c7d86bf1ca4a30f01c8 : SUCCESS in 54m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9fbae42af9f4a45be1120f7e520b8e9 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee15648a0264b82a065eefb0cd0fef1 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d6974c1c37e4e1eb77d96f82eb8912b : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0818f3a78eb74eccbc68e1bc94fbc390 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96c0e29dc6614af5a12f38caa4e89545 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aba9d29aa474112b960e0ca6adcf7ec : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fb1579dbdd643c6a297f5de3d4a016c : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0de3d3e09eb40c3a8437d7447f507a2 : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2054df26c27244ae8531aadee2906773 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3049e221f9ed49fa7afc42dd8f4817cbf5c17dfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:29:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0132703a9e405d4db622d803ec42d27121992d59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6809586a58bd422c9c8048f8543ab965 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63d28c08e12c33836935c2c945047cae107712ad","revisions":{"63d28c08e12c33836935c2c945047cae107712ad":{"kind":"REWORK","_number":1,"created":"2019-11-29 14:46:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/10/696710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/696710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/696710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/696710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/696710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/696710/1"}}},"commit":{"parents":[{"commit":"ede35f87f3a81eda25bb8795179134022c9c6643","subject":"Designate: pool-manager has been removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede35f87f3a81eda25bb8795179134022c9c6643"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:46:37.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:46:37.000000000","tz":60},"subject":"Bump OpenStack versions for Train","message":"Bump OpenStack versions for Train\n\nChange-Id: I86b65db4a7760c2f6d2494266389dce518df89de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d28c08e12c33836935c2c945047cae107712ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63d28c08e12c33836935c2c945047cae107712ad"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
