)]}'
{"id":"openstack%2Fkolla~696711","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ia1f17f3bffbd59ad99e46ddd100b88617737f29c","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ia1f17f3bffbd59ad99e46ddd100b88617737f29c","subject":"Bump OpenStack versions for Stein","status":"MERGED","created":"2019-11-29 14:48:12.000000000","updated":"2019-12-02 09:04:42.000000000","submitted":"2019-12-02 09:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696711-1575277401204-a4121484","meta_rev_id":"d2b2d715cc3437f806a2e691a9688edcfc5d2183","_number":696711,"virtual_id_number":696711,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 09:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-12-02 08:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-29 15:42:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-12-02 08:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 15:42:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-02 08:03:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-02 09:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c5ae6699392ec6ee286e6268e60560e8089fcfa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 14:48:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4986ef2dd974b05a49324fede1cfe35b8bb16a74","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 15:42:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c6627d4e716c8cef7b28541cae6acf7661d8953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 16:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8c7dcca5a9c14787b7a9d09547552f1a : SUCCESS in 59m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/487d4989606a4813a0065a9b2ebe2c28 : SUCCESS in 50m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/122df75d8c024737b9d047f60b0338a9 : SUCCESS in 1h 25m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d5cccd570b834855a67cd0413d6f5631 : SUCCESS in 1h 16m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ef6429fbe18b4ef4bab3cd8bcbbab260 : SUCCESS in 2h 02m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ce51cf44437c44498d0d9160ef794790 : SUCCESS in 54m 32s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c0a14c972c284083ade862cfac559a56 : FAILURE in 58m 46s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5fe95419625043e6ab90a6c015e07e5e : SUCCESS in 44m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e19277f998fd480c8af6104e745ccfa4 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94504f7da99247ddbf914bc59b641026 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94813a6c4483453194167a05025cb921 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f49da5806a4a269569b26586b5ca41 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/640a0f1f7af34e63bbf9da0a9f51bed5 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0b92267aae4ba394a1dfce2d95a6c0 : SUCCESS in 3m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1d1c10475af45b2a8aa4b6dc9903a8d : SUCCESS in 28m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ac2b40f76104029a15d29b091ff61bb : SUCCESS in 29m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50cf0c74f4af41d1b15747703ecec340 : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/859acd1f50d1457a9bc38f6ddda422f4 : SUCCESS in 44m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f604bddcd8f45e28efc9ec5f7c1ab19 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"12e383b77090355a09c0895aae12453c0b92be07","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 08:03:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ca2e60f212aab16d385ce9d5ed6e35c3a4001d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:03:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b556a188e17fe95ea579b676c30941fbdc3a2839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:03:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/134b4c17a24c4f64bacb5fb72e1e6b40 : SUCCESS in 53m 04s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f7bcffe2e8924524b8ad7e457aaad9ad : SUCCESS in 45m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c83ba7439e9846f8952448d2bc24d201 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1e379281194dc492b6229f9969c850 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/520a430432b84de0aa609e76810d4331 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1218b8b28c5a4258a1943f6aa1bc0b82 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c8bf865ec3241609d62260ba76e58c0 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10297fba13f84100832213854c41aabe : SUCCESS in 4m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2620b6c8ab645ebb26f98953d0d2399 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6530ff7f2618e944d7c81fc685dd4c506b77d920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:03:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2b2d715cc3437f806a2e691a9688edcfc5d2183","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:04:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25847e57a0ed4a5680cef33366a17d12 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b843a1eb18e5fcbef41656288b313a8e24dbfb0d","revisions":{"b843a1eb18e5fcbef41656288b313a8e24dbfb0d":{"kind":"REWORK","_number":1,"created":"2019-11-29 14:48:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/11/696711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/696711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/696711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/696711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/696711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/696711/1"}}},"commit":{"parents":[{"commit":"c0f40d85d0386abf87b7d293a01fdde0c3e32afe","subject":"Merge \"[manila] Fixes manila-share on debian/ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0f40d85d0386abf87b7d293a01fdde0c3e32afe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:48:00.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:48:00.000000000","tz":60},"subject":"Bump OpenStack versions for Stein","message":"Bump OpenStack versions for Stein\n\nChange-Id: Ia1f17f3bffbd59ad99e46ddd100b88617737f29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b843a1eb18e5fcbef41656288b313a8e24dbfb0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b843a1eb18e5fcbef41656288b313a8e24dbfb0d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
