)]}'
{"id":"openstack%2Fkolla~696712","triplet_id":"openstack%2Fkolla~stable%2Frocky~I9b9d077fe403d28d30a065bdc4be3d55a25f07c8","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I9b9d077fe403d28d30a065bdc4be3d55a25f07c8","subject":"Bump OpenStack versions for Rocky","status":"MERGED","created":"2019-11-29 14:49:14.000000000","updated":"2019-12-02 08:51:56.000000000","submitted":"2019-12-02 08:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696712-1575276716827-b0c15fbb","meta_rev_id":"be9701186c160b513e3a02e82cff9fb7535b34da","_number":696712,"virtual_id_number":696712,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 08:51:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-12-02 08:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-29 15:42:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-12-02 08:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 15:42:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-02 08:03:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-02 08:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c86bef435b1eece84af6f48b77da7f031e1b25fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 14:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac4ad37ced316776bb713fc42e7da0f386b5202b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-29 15:42:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8288a65e51750fe282ce4cb5cd27aab36d1c84a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 16:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6c2d61b2ab424703b9f7148585294933 : SUCCESS in 45m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/88c791027c2a49f78ed7e357d80a18be : SUCCESS in 56m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ef2c2cddeba14535a46c9ccde7d4c54b : SUCCESS in 1h 23m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2fa8d417cfe14d1db7bc7bd5572c701d : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f836f0c90a70488b8a91fca803205927 : SUCCESS in 39m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1e3b4b7e142647f6b7cf6e26e2328304 : SUCCESS in 54m 45s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c2932c92e0d54229bd977f7b847d3663 : SUCCESS in 1h 37m 56s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9a6bd2a5821a4734858413c234c7a43a : SUCCESS in 1h 12m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4417bf7dff9f4076a1550ffcdcc7d9a6 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20d5bda94b84753b333d27ed3312d43 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf339ab88a5b40fb8bf9d13434b9ad40 : SUCCESS in 3m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c845a37dc1254e5ca6956a78106ac6c0 : SUCCESS in 3m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/78120c0817c04a86bfde4f994af409e4 : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6557f90364c437ea6e8597db8e4498a : SUCCESS in 1h 25m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f980d75f03ca456cb79c5f7f311788e7 : SUCCESS in 31m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a56b2d052b1d48b8bc1bc6e3449480d8 : SUCCESS in 1h 06m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78e7e855bfed423cbca17f6c9d8e7667 : SUCCESS in 1h 58m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e1be71046b7ea5915c5fdb29792571a70741a58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 08:03:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b96477d703cd5557c687675b2f7aa14904f10ff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63b3c0300cb9401f732063f16badc99f1947e4f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:51:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be9701186c160b513e3a02e82cff9fb7535b34da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 08:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1a3b7876d65a432ebc738f4128e19a75 : SUCCESS in 46m 52s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/74215f5b1be5486d90c9bef411161b8e : SUCCESS in 43m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb54e3a28e845b883d77f9ebf568b42 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb06b5039a840a1bec3ad3299ee43a1 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e679f2efefaa4a3b8d1bda7a8414e930 : SUCCESS in 3m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bd79a6a3ddd342bc8434febc010ff4cf : SUCCESS in 4m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/074ecc5880ee49a9b32b1087c3c82718 : SUCCESS in 4m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3373ced616f3498caf53a585c922a15e : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77b664315dd8bb3a7262812488e7384ef188c4e0","revisions":{"77b664315dd8bb3a7262812488e7384ef188c4e0":{"kind":"REWORK","_number":1,"created":"2019-11-29 14:49:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/696712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/696712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/696712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/696712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/696712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/696712/1"}}},"commit":{"parents":[{"commit":"daafc9380b326cb04ced3033dd8686673df3dc0d","subject":"swift-rsyncd: don\u0027t recursively chown /srv/node at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daafc9380b326cb04ced3033dd8686673df3dc0d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:49:03.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-11-29 14:49:03.000000000","tz":60},"subject":"Bump OpenStack versions for Rocky","message":"Bump OpenStack versions for Rocky\n\nChange-Id: I9b9d077fe403d28d30a065bdc4be3d55a25f07c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b664315dd8bb3a7262812488e7384ef188c4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b664315dd8bb3a7262812488e7384ef188c4e0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
