)]}'
{"id":"openstack%2Fkolla~697087","triplet_id":"openstack%2Fkolla~master~Id927e5f5229eefee1114a8ec406ee8e1b089e6f4","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Id927e5f5229eefee1114a8ec406ee8e1b089e6f4","subject":"WIP: Set Apache to log to /var/log/kolla","status":"ABANDONED","created":"2019-12-03 13:28:53.000000000","updated":"2019-12-04 10:29:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3eacfa6ea91a966078ff675c80deb157054a41c5","_number":697087,"virtual_id_number":697087,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-04 09:19:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-04 08:21:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 08:21:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 09:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"446978cf49c9d0b34a6b3686e7d18aca2d89382e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 13:28:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af514f01c2366e3940ebf627719a0772962b913f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 14:06:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/87/697087/1/openstack-check/tripleo-build-containers-rhel-8-master/7568149/ : FAILURE in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"672a3fe01ef1b9f5b385672eb60bbf175f1fa00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:05:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8d461f6145754e05bd3a38fdfdd0fdb5 : SUCCESS in 43m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a2198dac5c364b4891fb026287b733d2 : SUCCESS in 52m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7821185a62af427da8cdc86103bf9d3a : SUCCESS in 1h 29m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6010c1f3c89a4ef5b282180ca4bb3933 : SUCCESS in 1h 24m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1ef497fae66d47b183baad7f4fea04db : SUCCESS in 1h 32m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/72c2dada8801466d82665eec5361c834 : SUCCESS in 50m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/45556bc153554352adf712d2c82a9f32 : SUCCESS in 29m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5f85fb921a824c958c07d2828bc12219 : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1009baf5ce2b4c7a901b5769179422be : FAILURE in 29m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0da028fd89648e08afe9892808ecb61 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2e821254e34016a1e51445d0fb423d : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c153cba71e34bd6acb9a3bf27f7a5da : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91fd7223e291435796cb8c4ace315f98 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e0a13d3db04077897ba721b502201e : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab9ab77bb9b84f54a7485a6347438aef : SUCCESS in 42m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d8238f414274cef93c26d8cd66cbcc2 : SUCCESS in 45m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1a00b36492741f7b2db08e84a323a1f : FAILURE in 28m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42c610016f30497e9c754983db7b5eec : FAILURE in 44m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ed3e3f5e0724cb5b8f96bdbc2e610a2 : FAILURE in 1h 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"692ea74467d169795012c026e508332a304bc510","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 18:02:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16a090061bffb937457f9b86990fe62da0f07215","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 18:07:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"06c0357f17ead5ea843734f0d2df6b29d584f3f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 18:23:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/87/697087/3/openstack-check/tripleo-build-containers-rhel-8-master/c3e3270/ : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c4c8e54981e895d62df673874134eac74dc27ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:33:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8053e0e3fc254fcb92845430cd51dd42 : FAILURE in 6m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d0c09145aae74c7299fc8ae9aa27a6b7 : FAILURE in 6m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0d24b4e53bd24a9abdcc5ec3b3a96aed : FAILURE in 9m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d6581684588b434b94b12c8d90e32bee : FAILURE in 13m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7594c444df9849379c4fba4269a1f579 : FAILURE in 11m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5c70f9fa1e354e95b4dcac9ec20ebd23 : FAILURE in 18m 57s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/3dfcf305efd14169ae8f23fad1688dfe : FAILURE in 9m 29s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5d3c90aafb2a49b0b8f5008a6d33568b : FAILURE in 11m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c3a6f6f06648452d84d9b8bc1fc1ad2b : FAILURE in 12m 56s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ae80cbff2cb441109822afb2733c1259 : FAILURE in 13m 33s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cf0b1134cf71431c93ef4f92dc5651dc : FAILURE in 16m 11s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/0d3c53cd962e4348a0f98da6659a9b68 : FAILURE in 9m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c928b6136356461fa2dc72e1114e040d : FAILURE in 5m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/07fd45dcef4347dd9804174c6154a2d6 : FAILURE in 5m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/96ec61fc252848e48b8ed894889fe253 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5474589fa50d4ee197f50c68d1ddcbd5 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49337d0a0798422bb4684c92a6f9cdaa : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/123717b9a517478baedb327b042984ed : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/606d8916e7444583979b3fb057c1e725 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc3f3c28d1f4a3f92fcd4a6a0b093a5 : SUCCESS in 7m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e6afcfa85774464a6d6562509798211 : FAILURE in 3m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8ccde1eca434c69910be359a9996a51 : FAILURE in 4m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/635cdec7d2c048a0af2fd79138eb2f2d : FAILURE in 10m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0a8246aae0b4f9db35db3d484626331 : FAILURE in 8m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f781ec93b814d5fb86c280399d672de : FAILURE in 7m 19s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c208df5e36974725ac224864d50105e2 : FAILURE in 9m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/647c2656b55e4f7cbab21597d4a41db9 : FAILURE in 10m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/bf6034ea6d3e43e6b4fa8141a2dcbf29 : FAILURE in 19m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e9c033ec6798417e8a85d83e62864812 : FAILURE in 8m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a9506f1a34ee4429ab6759eaf145673c : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89b0d159efe519264abbdd1478af75c499be1795","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-04 06:11:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"024689f6c00c160a481330e5f3c8a87559fcb85f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 06:40:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/87/697087/4/openstack-check/tripleo-build-containers-rhel-8-master/44a0406/ : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"7b71ed89bf5eb6260edc6ff8176ac6ceb5348cb8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-04 06:52:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1259a7049b4d74731e14fd45441687eca56e8369","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 07:26:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/87/697087/5/openstack-check/tripleo-build-containers-rhel-8-master/7195a1a/ : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":5},{"id":"55ba8f6c760916c4d18fac6f2decd9d0556ccbdc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-04 07:34:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2472ae74b10d920ca2ee34ddc7f98cc4ff7f2b19","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 08:21:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/87/697087/6/openstack-check/tripleo-build-containers-rhel-8-master/8c9658a/ : FAILURE in 39m 20s","accounts_in_message":[],"_revision_number":6},{"id":"04cbb58eb3ca2ee0c8ff5ef36fae4ee5a027a0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 09:19:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6639cc60700a41afaf6644de4e84d5fa : SUCCESS in 45m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/297431387d074b6aa83ac9195aaee76a : SUCCESS in 51m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0ae73128315244d9b2d6dd2166d05d12 : FAILURE in 57m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5812407a623b4d189b750cee3127e8b6 : FAILURE in 51m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e4488f797be44fabb3204ce4f2eeb56b : FAILURE in 1h 43m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2bddf352d5ed4b16b0896b5421010e48 : SUCCESS in 47m 00s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2a1a23c2281f49b38d568e90b43cb19c : SUCCESS in 41m 54s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5f788399d280430e903b6fb0639b4947 : FAILURE in 1h 22m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8ff490a3d53049fab277a70b8ecd4c90 : FAILURE in 1h 27m 00s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b4130a1c5d8345d8be7ec6bda0d982a0 : FAILURE in 1h 21m 31s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/41b8078cf1e54a0fa505d24b1563a4c9 : FAILURE in 1h 18m 14s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/152b73287e9f4f2887dd324abf46e01e : FAILURE in 55m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d22b14a0fab5419ba6cdc0cd05d67482 : SUCCESS in 23m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0789f6c41279449aadea844bcad5f4bc : FAILURE in 26m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/689f2a9645284f91a62f213b878c2247 : FAILURE in 41m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25ed6b64ec3648c6b2f8a6768413bb82 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e6b86d6172a47d79e4137b7efa5422b : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131250f8cf474bd882dea16f2e54ede6 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79f94e4a46e04d10b1fa36b8f10f6231 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c159592fe78460aa10d695dda836273 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d14bdee32cce49eb9964f02ab47fe593 : FAILURE in 29m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/179e735c092244249bcd77b94dabd4c3 : SUCCESS in 37m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3096c379e3f4f879e37e0c462e12c9a : FAILURE in 35m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b366f9ffd88a4f69af54409b6db5da14 : FAILURE in 39m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/33bab9e43a154462a31b3d08b15a6289 : FAILURE in 59m 07s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f5cd1be84cf54fa19599b6d2a74ea7dc : FAILURE in 53m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e3a244a5ff81477098ee64870a5c835e : FAILURE in 1h 02m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f2e8890755cd4ba99c48917c9b63829b : FAILURE in 57m 36s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4e9bf56843d3477098d1920ce76389d0 : FAILURE in 50m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/26cceaf347564ac9866f99892b27421e : FAILURE in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3eacfa6ea91a966078ff675c80deb157054a41c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-04 10:29:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"85b618b1f1c1419134bf9a3bbba04636a2bbc86b","revisions":{"581c58323183ff4c881248dc8c2fb555c3046bd9":{"kind":"REWORK","_number":1,"created":"2019-12-03 13:28:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/1"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:28:50.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/581c58323183ff4c881248dc8c2fb555c3046bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/581c58323183ff4c881248dc8c2fb555c3046bd9"}]},"branch":"refs/heads/master"},"b89d4fee65fc94c437f3cd7e6f778a9f5f87fbdc":{"kind":"REWORK","_number":2,"created":"2019-12-03 18:02:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/2"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 18:02:01.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b89d4fee65fc94c437f3cd7e6f778a9f5f87fbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b89d4fee65fc94c437f3cd7e6f778a9f5f87fbdc"}]},"branch":"refs/heads/master"},"9cfa9f3bc318846e4835c33119353e7ecac7f2ad":{"kind":"REWORK","_number":3,"created":"2019-12-03 18:07:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/3"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 18:07:24.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfa9f3bc318846e4835c33119353e7ecac7f2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cfa9f3bc318846e4835c33119353e7ecac7f2ad"}]},"branch":"refs/heads/master"},"d483a901b558000ddb06fcd174c2fee69b8e5215":{"kind":"REWORK","_number":4,"created":"2019-12-04 06:11:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/4"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-04 06:11:05.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d483a901b558000ddb06fcd174c2fee69b8e5215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d483a901b558000ddb06fcd174c2fee69b8e5215"}]},"branch":"refs/heads/master"},"8fe828926c1d5ed389df7ce3fc5ebd74289b9126":{"kind":"REWORK","_number":5,"created":"2019-12-04 06:52:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/5"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-04 06:52:40.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe828926c1d5ed389df7ce3fc5ebd74289b9126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe828926c1d5ed389df7ce3fc5ebd74289b9126"}]},"branch":"refs/heads/master"},"85b618b1f1c1419134bf9a3bbba04636a2bbc86b":{"kind":"REWORK","_number":6,"created":"2019-12-04 07:34:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/697087/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/697087/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/697087/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/697087/6"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 13:27:45.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-04 07:34:43.000000000","tz":60},"subject":"WIP: Set Apache to log to /var/log/kolla","message":"WIP: Set Apache to log to /var/log/kolla\n\nChange-Id: Id927e5f5229eefee1114a8ec406ee8e1b089e6f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85b618b1f1c1419134bf9a3bbba04636a2bbc86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85b618b1f1c1419134bf9a3bbba04636a2bbc86b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
