)]}'
{"id":"openstack%2Fkolla~697088","triplet_id":"openstack%2Fkolla~master~Ib56e62d1fb4d0fc4a6c627b87a929be0bc614d1e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ib56e62d1fb4d0fc4a6c627b87a929be0bc614d1e","subject":"Install etcd3gw to fix Ubuntu binary tooz coordination","status":"MERGED","created":"2019-12-03 13:29:47.000000000","updated":"2019-12-05 09:30:48.000000000","submitted":"2019-12-04 19:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697088-1575487642342-612455a9","meta_rev_id":"e868b34cb706568cb1544027c121498d117dfed5","_number":697088,"virtual_id_number":697088,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 19:27:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 14:07:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-12-04 08:53:35.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-04 17:29:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-12-04 17:29:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-12-03 13:29:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 14:07:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:53:35.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-04 17:29:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-04 19:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b57ffe01eb146acd479c4eace0b834a56b63d425","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-03 13:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d00efc40db6880d48bd91b21572f14edabf68f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-03 13:29:58.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e48709922dfa3ead037b6fd73ed1e7e7b59f4372","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 14:07:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/88/697088/1/openstack-check/tripleo-build-containers-rhel-8-master/f81410b/ : FAILURE in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"202c76b2ed2cb01c578561abf8230b31409b3763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:51:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ef38015f94f54341aa4f9b7cbbd7d2f7 : SUCCESS in 32m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6d6e3f33de5c40489125e299d6e4fba1 : SUCCESS in 44m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/efbee14ef3894d49a6714d1bee3de141 : SUCCESS in 1h 31m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/24054af30db341808611ec93aaff6c04 : SUCCESS in 1h 42m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/598e357a7c8c4bf28ee2724eb86e1a8d : SUCCESS in 1h 23m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0dc614d6633f4757bdffe86a1db24ba0 : SUCCESS in 58m 45s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2a2274ffeead4f79804dd07e46a437c1 : SUCCESS in 51m 24s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a9aebd202a244993a515d371470a1b24 : SUCCESS in 2h 08m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/40264af9c8554454af45f037004c0df3 : SUCCESS in 2h 10m 22s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7300ce69c5ae4fbbb3dc20f1679ed303 : TIMED_OUT in 2h 03m 19s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/0b2a86394be445cd9c6c30474f4ec2fd : TIMED_OUT in 2h 02m 34s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/2fe3dc3f99eb489d81118afa00b4c9dc : FAILURE in 11m 06s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e498f36875e74afcb7852bd734949ead : SUCCESS in 45m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2a9f62a0cb164d53b175a235d98a0cd9 : SUCCESS in 44m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/99c7863d50314223bfa8b30b01a7d710 : SUCCESS in 1h 10m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e91ea719316844d2b55263551bd87b8d : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e6a81e395b4106938bbfa56f93b89c : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7e974eb87a471aad7c8829edc3a849 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c9f27f454c4443e93b66428bb4d531d : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d669ce9fb94d4daf815b2e49485fc663 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf2eb34d67e04155b022340e62ab4803 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0b92608360b430dbb8941b7be947720 : SUCCESS in 26m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0421958fb8c046bc9d74fb1174d9c66d : SUCCESS in 31m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a84bd4ec2ea4bf6a38fce85ceea465b : SUCCESS in 1h 21m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1d53620373e46a98e973dee66742216 : SUCCESS in 1h 26m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d23ff49e65554384aad27d7cdc4e4e52 : SUCCESS in 1h 33m 54s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0d92c4fffe55415686f99e60538b2ad8 : SUCCESS in 1h 22m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/bd3563b8650544d39eb0516658264441 : SUCCESS in 1h 24m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f70f55eb25184a4e83df2b4442b106cc : SUCCESS in 1h 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4f726b11cb1e4d6a8f3f5eee0ae9742a : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0e32185403484054945184855092c921 : SUCCESS in 1h 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a159437afe8a3084bbdc5d0fca91eaba69cc8b8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-04 08:53:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb4b8bff26dd5541922190aa7517cff45da8ba66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-04 17:29:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb05f50861c0e6c29881021c99be439e0b4e4c8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 17:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"042d843a2c10eba36480c80c37e791a3d413cea8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 19:27:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a842818b6c544c712f9095afb455728493e2dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 19:27:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/14aba98bc57d4d7d9c4bd714ba817349 : SUCCESS in 34m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f0aad37e555e47f59051157a249108b3 : SUCCESS in 1h 18m 34s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/21ea32e33f8a422f94d8adae8659e507 : SUCCESS in 1h 53m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/128f898f5a0a4e0bb0fe6d80f7137096 : SUCCESS in 1h 07m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/950372e5634c4699a5eef46676d120d3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a11f5699a34455a1e78fce8071fef3 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10212875a35c4a2eb82f33aa1942f1d7 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efc6624af79246d5bcfccacbdf482403 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f239e9fd134faf8e7c27b63b853604 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c755ed7826ba4646af17869bbbfd8323 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1dff7d2709d453ea0dd0df5dca97d06 : SUCCESS in 25m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/59ec1dc176084e55a8c3885692827188 : SUCCESS in 1h 00m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47c69c08f2284faf839c1dca4caad0c2 : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":1},{"id":"69972c5e6e012f86ebda75ea402444375dfe6b45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 19:29:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9578b7e9c434a369ddff6bfd9173f2a : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c16a9ffa444845df8295ac2fa3f80fb7 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e868b34cb706568cb1544027c121498d117dfed5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-05 09:30:48.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 19c38d9c9d6416b7859037396484fff2e587ec4b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0186c5e3cafeae0bb1c39bdef0df6ade827552d5","revisions":{"0186c5e3cafeae0bb1c39bdef0df6ade827552d5":{"kind":"REWORK","_number":1,"created":"2019-12-03 13:29:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/88/697088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/697088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/697088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/697088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/697088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/697088/1"}}},"commit":{"parents":[{"commit":"9420c991ad729a11c65169925aea6d6fd4c5372f","subject":"Merge \"Added Neutron Mellanox agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9420c991ad729a11c65169925aea6d6fd4c5372f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-03 13:28:26.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-03 13:28:26.000000000","tz":60},"subject":"Install etcd3gw to fix Ubuntu binary tooz coordination","message":"Install etcd3gw to fix Ubuntu binary tooz coordination\n\nChange-Id: Ib56e62d1fb4d0fc4a6c627b87a929be0bc614d1e\nCloses-bug: #1852086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0186c5e3cafeae0bb1c39bdef0df6ade827552d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0186c5e3cafeae0bb1c39bdef0df6ade827552d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
