)]}'
{"id":"openstack%2Fkolla~697354","triplet_id":"openstack%2Fkolla~master~Ide4868d031bd84b1a6935aada7d0e7627f11bf37","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ide4868d031bd84b1a6935aada7d0e7627f11bf37","subject":"add more tests for enable_repos template method","status":"MERGED","created":"2019-12-04 19:44:24.000000000","updated":"2019-12-17 22:34:57.000000000","submitted":"2019-12-17 22:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697354-1576621931643-fead5b33","meta_rev_id":"70ade94e75c72f6d279bed2445ec6241f30c0dbf","_number":697354,"virtual_id_number":697354,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 22:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-17 19:37:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-12-17 19:11:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-17 19:11:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-12-17 20:04:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 13:41:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-16 17:02:11.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-12-17 12:46:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-17 19:37:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 20:04:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-17 22:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7a1d6cb3592393f82be01a3e2d6c165bb95dc64","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-04 19:44:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"631756beff2e877a09400fe9d96e07a09e7d6679","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-04 19:57:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37b283f9936b901fe4165ab23ce12e885f5f12b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 20:33:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/697354/2/openstack-check/tripleo-build-containers-rhel-8-master/328f5fc/ : FAILURE in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c4490365e1769421dd143ba4bb1ac7f7f5c64350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:49:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c7fa13885c044689b0c7d583608f58d2 : SUCCESS in 30m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9038854f3b0e40539fa8eac140da3063 : SUCCESS in 49m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/044eb0e2570e4e0a9a575b5f5108f905 : SUCCESS in 1h 22m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/586583cd3389404c82976de6fb61281c : SUCCESS in 1h 35m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/96fb08b737f14ba680f4d88fe435226b : SUCCESS in 1h 49m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b30f3de070134be69adc7c540d822d2d : SUCCESS in 59m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0861bbaaa5424619a04a022afc0d287e : SUCCESS in 35m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41b2650faa904435bb292cbdaab25009 : FAILURE in 21m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8249e2ede4e647c7b3582fb6b8f18093 : SUCCESS in 56m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818523e92c204e8cbcc86674d56b6d95 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d1cb6f224a41da8e1a0be5f81170f0 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f03e9e6858f47e4b0c87221a12b2e12 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5819085f92743f492a22bc4c2682e46 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418ad69bd9e24d21b54e712c835758da : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29b8db1d380d4b6eb615e3541e674e97 : FAILURE in 24m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/820848e0e8a9488fb0707b115a5292fd : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dff4d0143ed14e8baf765f213c8b04da : SUCCESS in 1h 00m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b05939a416c47768e48649daeff8077 : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/365b9327ceee46fcbfc3916c37c4a954 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"62d511027713f8653f15fc23d165d9069261ee8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-05 10:14:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"507374014c9ef4e1378f02ec4a02d81bc40c2a85","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-05 11:18:46.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a9eae8c561b0f9e1bbd30185b8882c8068ec01c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-05 11:18:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48b8c76137d56335131caa9dcf08df980a3f6679","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 11:50:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/697354/3/openstack-check/tripleo-build-containers-rhel-8-master/2b0c91f/ : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ca9509ac0bfe0391ccf2d2096ecc11db696ef4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 12:39:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/826d9095b2604115b5ba6d7d3fc34fd5 : SUCCESS in 27m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d7c50b277bce4f8c8bdf78a5def03b76 : FAILURE in 35m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/59294b955cd249718c610635822c8568 : FAILURE in 41m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eec5156d44a24cc68cdbfc6d22a7634b : SUCCESS in 1h 20m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4d063df99e0345ddbebe72b1d898bc08 : FAILURE in 35m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8980101ae5454b3681b780ec6614b2b7 : SUCCESS in 58m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cc73523376724ebcae714fe1c4b8ef03 : SUCCESS in 22m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cf85080e94e54ca1af7bb238dd891cab : FAILURE in 17m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80de43a2165540a39885614618257c10 : SUCCESS in 57m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80ce3c4e58cd431391fa3b7e6f10e5e6 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cd5c418dec43e0b6817f364abf5c40 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8986c5bfb624db59d73de05164546ab : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf26abba47084df4abb70659b45064b9 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2112e1ba0e96403aa92d03190177ae75 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/571c144777b948a7a21636d52d36611d : FAILURE in 23m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e65655dfbd745d38d0c7372ac08a8b0 : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34fa25480b79477192e8cd8a1558cc35 : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0bb2e6d16fd9444192d626c16d718db2 : SUCCESS in 57m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f6d14519ee0445c8fdbfbc6ed65af1b : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":3},{"id":"db1d15ce7841d95a679386511e20f008d4a9bc38","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-06 09:31:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"984a899dededdf9f1d3ce3dcdc05e3cb08d0316d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-06 10:06:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/697354/4/openstack-check/tripleo-build-containers-rhel-8-master/301b454/ : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":4},{"id":"ca0d40985209c43826ce14e9067e794aa0619d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 11:05:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/64eed9dabfdb49a38f94e8439a19382f : SUCCESS in 33m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/59e78d3310ab4271a53f279f603b3275 : SUCCESS in 39m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d3a1d94ca47e4bac98b0d0fe936aa951 : FAILURE in 57m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e0108c31bba94c71a6c8988dd9cc153b : SUCCESS in 1h 15m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/06f920c2797d4df6974eb7057c44224b : FAILURE in 1h 32m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/19d612045ddd473f85665829c8a6cc32 : SUCCESS in 59m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0157aa7e37b340b8a5655e78000e7966 : SUCCESS in 28m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/02ed5adfc5804a47bb2f44677c9af75e : FAILURE in 18m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8a702b867a3e4c0b826aff82dfe73597 : SUCCESS in 1h 03m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/550725ce4ba14928a5f6de5cfe1003ff : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8c508cdbd64529b3d120ef7d76c2df : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d5c8410b6a84e5dbc4ab351258ed83f : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b2c4911fc1145da92138aae9a629056 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa55313903e4658bc835f7e0f0bae72 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/795689be9bca4a7abda247fcc4078b7f : FAILURE in 31m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7744bc2c8eca44fe9db11a16cf9bbd67 : SUCCESS in 33m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53f24b05716048a19387194496508e0c : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07824ac373c743a29103aaab1752df1e : SUCCESS in 56m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/77d344ae9d3241acac44040abf0e3350 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":4},{"id":"656fbcf76f0449ada43aad8a9ceddcb47236f36a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 10:00:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"aa983931305a564d74e5b05945d2a3e94882062d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-09 10:42:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/697354/5/openstack-check/tripleo-build-containers-rhel-8-master/56b74b9/ : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":5},{"id":"72f1bd5cad5229f503dddff1c3c22b6cf3751a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:41:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e5ddcb1149094bf5bab630e162aa0715 : SUCCESS in 28m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/35fea7460db44f7bb8b28f40f9604f87 : SUCCESS in 43m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7d4d32aa8c234e41ad913df78b92e759 : SUCCESS in 1h 13m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0be5acb8f6214d329cf99fa3d30d5550 : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0c208efc6dc5442eaf6847d9b031da8b : SUCCESS in 1h 36m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/26d5f10380d241888d05544ceb06deeb : SUCCESS in 43m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d6f8b2c789e4447b95f8c4716bc5f3c : SUCCESS in 22m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d590fe6eb39b4d6d905bdc855805b39b : FAILURE in 8m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6e0eac09eef45e885630ff87b3a7708 : SUCCESS in 59m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fa4345e7c3f4ec4b18471595e9a6517 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e330d281684a9dbe4a9081b5af8aa3 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cb267b76ffa44bb9c24ed0f552907f3 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22d5d8bf6a7c4f51b489a298a97afb61 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7869c4f1739645e3b435042131bf4404 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d39ef475d0e54b018817df661f695084 : SUCCESS in 24m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0ab85fb86ed4500a96f0f7a1a0d89db : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d8dee7cd6b848b19bfd03ac5d145f40 : SUCCESS in 50m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4082fc64887a41f4b3e19f4fc0fda294 : SUCCESS in 56m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b24fc333a474a1892222def00b0fc05 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":5},{"id":"86aa54623510db29d1e76b9d9ec2489386657a51","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 12:15:39.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"53e434c0e205a5a71cff41a3a5c4c79af6ad852d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 12:15:46.000000000","message":"Patch Set 5: Code-Review-1","accounts_in_message":[],"_revision_number":5},{"id":"30cf4caf3455a07ffeb124996b5967d9aba84278","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 13:36:34.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"64cf6931306129296932c1c98964caa2ddc58e3d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 13:40:15.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"09affa5218eaf4152708bd2f14a2b1e99ca15df8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 14:59:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9dc3d9b74ba63d06d712041015a85e80a175f711","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:17:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09f23c2fe3fb419fb514f9c87c77471d1bd5a83a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:17:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec8b7c4710454428955c1fd837e7cc999a0de402","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:18:37.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"916d54ace9136ada5517de9f30b117445da58cd9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:43:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"65249f9c39057f7bdeef8be1ec092a72c7a6c8a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-09 16:19:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/54/697354/8/openstack-check/tripleo-build-containers-rhel-8-master/46c33b7/ : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":8},{"id":"aa42dfac82a3256ff18b42e138417a72f431f280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:31:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e1f1c1ab296f4e04abe24d47966e5253 : SUCCESS in 40m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3fcfda3ed73a4f96bcf086f92da2e0b0 : SUCCESS in 54m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8bedfd9e56e04b6984fad914ce6b3a5f : SUCCESS in 1h 29m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f82962d10bad49b599bf76b08abbcc3e : SUCCESS in 1h 32m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4af9355ac9e7490da632e5b77e6263d6 : SUCCESS in 1h 42m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4e7f3cc20c974b409bf8bd1bedae7125 : SUCCESS in 46m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/78f6aba84d81459ea692d13d4d065397 : SUCCESS in 28m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e4eec0a9e2ea436ea38a9ce72afec161 : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/99d7fd193f384553b027f9af90eeee50 : SUCCESS in 1h 01m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16a9f02e6ea945c8903ed735f7110704 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14863f1c37ed4af2b85fdb4e4f6240b5 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3741ce024f3b401fb8c3e530b7fa4eba : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1350b8558ba7446a8b95dab0165a5674 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a812d4c1d492485fa1a1d519c90353a8 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1097e5036feb451596d971013678610a : SUCCESS in 24m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe5a56bb547e480fa6e11546d4ec4de6 : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e9aeac11dd3407e94ac33bd6e2fbcec : SUCCESS in 1h 02m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af80bb335a6c4707a0a923ebdb05d1dd : SUCCESS in 1h 00m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0eb221a07ba944d1a4e394e5f698b1be : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":8},{"id":"01399fe9f18dac768d9d122640913c928307c175","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 11:11:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c3a60b9d55c07f297207b3ed2fe2331686f578f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 11:13:00.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"d10dfc0929700e361182ad1123026855044e5e90","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 08:57:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"339ceb6a235f5daee069ab2ef5b737b0e235a5cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 09:28:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/271afbf024c64ad486896e98694e9d39 : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":9},{"id":"d00f57669a9a8fb6754fe1059646de153b95b17f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 10:01:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a2db74b69624978bd443663d213b03e : SUCCESS in 1h 03m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"101934ee1ae8e7696977ee57b745e88540cf6e0d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 10:24:50.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"75c702cebf17dc6be2bfd2e22373d414706253f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 10:25:45.000000000","message":"Patch Set 9:\n\n^ even though gerrit says PS2, it is PS9","accounts_in_message":[],"_revision_number":9},{"id":"d8dcab6dd69d281bbec35a11e19af087efa3dbd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 10:36:07.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/78b9ff19b13f44419de678bfea8ed980 : SUCCESS in 26m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/288f1ecbdee24cce92349e03b7290116 : SUCCESS in 33m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/292462e169874461a40c7e15914dd2c2 : SUCCESS in 1h 11m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f83205a0e9ae4fa48e1d6d892bedf72d : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bf90d0c340444d7a8309b020c833e179 : SUCCESS in 1h 33m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3ffc790783114b95afb1f9190c4df8ac : SUCCESS in 52m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/890c058a54c54ae0aac420fd74bfed27 : SUCCESS in 26m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c6e5f2663a04cabaee537f717f46e4c : SUCCESS in 24m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ce9f3baa41540a99e0d9ff1e46d4a35 : SUCCESS in 57m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f37a6a96a04672893b9a8c2d9177e7 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed64f013ed3045c7883fd04b6d1004cc : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c95ad7dc534be8a845e11f7d51ef21 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84c79281f3e540dba285a7053e0d2928 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5418fd8b317491f919cdd3721137cdd : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/623faf98444e4d6684d883e9a8f69126 : FAILURE in 23m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e695fcd640f24fecbdc130d089e8847c : FAILURE in 37m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2c88c05e7984139b6900d804b23207c : SUCCESS in 59m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe22c0e585b4484292fddbc223b31271 : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e5c9d238ba2493fa9a8b737de4fba1e : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":9},{"id":"9805e8528a3b4c18c7518b3e9aefa1ddba58c2e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 10:37:32.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c5da5444429e6402c5543f6fa743734b20662ce4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 10:41:59.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8406332d63316fb0f33d39dbe6569382b6309743","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 11:02:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1689ea65f3de6819177ad24478aaee538ccde5e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 11:21:43.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3a29c470889fcf794466b5e03cadf6ec8971cba6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 11:57:06.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bf5e4ceee6bf4a61fe540ffeacddba5065690cff","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 11:57:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8aee1c0578424d284ee2e7f13184294df6b3cb97","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 12:26:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c31e18974ab5411f90365f10bac2b5fb : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":10},{"id":"7b2ca1010a67fa1d53ed41e7420a53c80393195d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 12:46:27.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"c9d2607dab894dde093e6273671daa4b076d4e26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 12:59:00.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2f7d3b5bce382c2af742de47cb833f3453ad36ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 13:22:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84ed97c857d94221bbea053c8d9da926 : SUCCESS in 1h 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1afe55bd2d02e1a6856119003412b7a7bdfe78c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 14:14:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7289eb1c42a34115850bf012e76b4554 : SUCCESS in 38m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/97f6b8a95bce4dbe8532789ab3796b07 : SUCCESS in 44m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4e7b4ac4410d419d87af235b1660a929 : SUCCESS in 1h 28m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a7f107146e9a4b08b7e2ca95e5309681 : SUCCESS in 1h 27m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ca5471977984f06a042c16bebb26871 : SUCCESS in 2h 05m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0769848df8fd4799a09bde779499a695 : SUCCESS in 57m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b15393bc487f4e638a9d903b8d8e49e1 : SUCCESS in 35m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/afb501c1bb794d488047e8bdb22519c7 : SUCCESS in 37m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/402f87adf5c64b769a0fe453c8760782 : SUCCESS in 1h 06m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad2717bfd23d4fb4b10b74ee16d14d3c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225db79500364ef3a72d848ca99e79d3 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/980ec95320cc4cbf989995d4c26a01c8 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b1aee64553c447390d260800dbce9fa : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7825062abc74341987dd479f65ede55 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96b8846df2b147d1a3c7a7bf2f9a290b : FAILURE in 36m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10d4ecaad11f4c4aaf9b494e675f4bde : FAILURE in 47m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bde43713cd8a43a3afe0a7215a30a963 : SUCCESS in 58m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4d253b2231348138b5a79ebf03cc763 : SUCCESS in 59m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0bfa296438e4aa1b29dd8b5eb0bbfd5 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":10},{"id":"23bd050fea28e27fb3ee9676ee06ae10d3f3a173","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 19:11:42.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"904c884e18aeadb3d7ab99a210ec0b194daef0ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 19:37:19.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f38cb665acea4152a704888186b9bcbe : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":11},{"id":"3a6994302542910e49a7c055f44be270394fcc01","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 20:04:33.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"4a32a509417b57ea82b8760c2592292e9a62d870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:41:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e772225948846dca047c055f9afd34f : SUCCESS in 1h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea1e542e060ce04d0fce29522c448c529c70eb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:51:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/657915b9970b4c62aa3c5b01b48f1ddb : SUCCESS in 30m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dc22c78b0d04451e82ea8632acb28622 : SUCCESS in 39m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1599361d3f6e4d1ba52fa70568f25ece : SUCCESS in 1h 17m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0f8cf6e2a99c43909f43aaed323b0da4 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9ed46952eb0046ab81d5aab69fa6f1d9 : SUCCESS in 1h 34m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6bb0b593a75741ac9d0001080f8966b3 : SUCCESS in 51m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc7bce88f4c74b42a7f4ee087e788d96 : SUCCESS in 39m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/20dc2e29125b4fada8b88c57006ba438 : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4726319d08194ddf8c9a61ee9fdda5d7 : SUCCESS in 57m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06552cde4a724c309cea07bb4e13467f : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a015c9d9ffa470e85fb301311784504 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aedb15471e964c339d5ab1915c812443 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37ab4636c154424caf41e7b2d6302d10 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9021617d3b51493dbeeb31370bac7415 : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5eddb497f0d043cb9ea5052f8ce001ba : SUCCESS in 25m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd142e17fd704125a8536ab8ca9ffc9c : SUCCESS in 40m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07cd193a147041b59398fda6ba20d4cf : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e825577d04c46b497590e85ae170f1c : SUCCESS in 59m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0727c0726fb4339afb1f13aace9009a : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":11},{"id":"548b28f36ba129375ecb0c58add1614e3c7f99c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:54:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"e24d132707c78ec85d261185bbfcf863fa784d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:32:11.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dc479247abcf4164ae9d538953148012 : SUCCESS in 38m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/31f175af60b34837a0a2e4586d97168b : SUCCESS in 1h 23m 08s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/00fa91e5f6a6446daf668b40b7252f2c : SUCCESS in 1h 36m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1ff861a802d74899b08d65a61adf3a8d : SUCCESS in 47m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af42d9434f014bca99679ad3b3f04b23 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c14a9c4cdf40ffbc39df5bb325b987 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c220e07044c49c58c089b3cf049f130 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d9e598a45944af2ad5ed498735ebeca : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9676c9f22954c06a96505fe55821e73 : SUCCESS in 8m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76867eae774942bf8301bed1bc98cc06 : SUCCESS in 44m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/135045bd9ffd4dc68beb42ab27eefd91 : SUCCESS in 1h 00m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64d001877bd1484499be3cccb053c3af : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":11},{"id":"aa19efa48a42eca73823a20024a42ad812a18585","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:32:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"70ade94e75c72f6d279bed2445ec6241f30c0dbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:34:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9664bd9eae47c4ba8d461cf15de31d : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"540b0cf6d7da4f7a85c54af847d4f519471d6e27","revisions":{"cda21502f267169320dd4f5657a6df052d7392a8":{"kind":"REWORK","_number":1,"created":"2019-12-04 19:44:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/1"}}},"commit":{"parents":[{"commit":"324ea9e2baec0a135fc0096922ffd498fda53eeb","subject":"Merge \"Install etcd3gw to fix Ubuntu binary tooz coordination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/324ea9e2baec0a135fc0096922ffd498fda53eeb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"subject":"add Debian test for enable_repos method","message":"add Debian test for enable_repos method\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cda21502f267169320dd4f5657a6df052d7392a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cda21502f267169320dd4f5657a6df052d7392a8"}]},"branch":"refs/heads/master"},"ac3261c4ac60f0afdf9dcf9643291c92fd1b4a82":{"kind":"REWORK","_number":2,"created":"2019-12-04 19:57:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/2"}}},"commit":{"parents":[{"commit":"324ea9e2baec0a135fc0096922ffd498fda53eeb","subject":"Merge \"Install etcd3gw to fix Ubuntu binary tooz coordination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/324ea9e2baec0a135fc0096922ffd498fda53eeb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:56:48.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- empty one\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac3261c4ac60f0afdf9dcf9643291c92fd1b4a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac3261c4ac60f0afdf9dcf9643291c92fd1b4a82"}]},"branch":"refs/heads/master"},"2fd5e7c69ca06b3c0dc62834d01226c54bec0b84":{"kind":"REWORK","_number":3,"created":"2019-12-05 11:18:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/3"}}},"commit":{"parents":[{"commit":"324ea9e2baec0a135fc0096922ffd498fda53eeb","subject":"Merge \"Install etcd3gw to fix Ubuntu binary tooz coordination\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/324ea9e2baec0a135fc0096922ffd498fda53eeb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-05 11:17:42.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- empty one\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd5e7c69ca06b3c0dc62834d01226c54bec0b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd5e7c69ca06b3c0dc62834d01226c54bec0b84"}]},"branch":"refs/heads/master"},"1ba949dc9e225d62e2e53c853c79c05ab3ec1b61":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-06 09:31:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/4"}}},"commit":{"parents":[{"commit":"df987c9d3fca40e0a7529e48a3626dfdd58e2168","subject":"Introduce macro to remove Py3 only plugins on CentOS 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df987c9d3fca40e0a7529e48a3626dfdd58e2168"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-06 09:31:17.000000000","tz":0},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- empty one\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ba949dc9e225d62e2e53c853c79c05ab3ec1b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ba949dc9e225d62e2e53c853c79c05ab3ec1b61"}]},"branch":"refs/heads/master"},"517fd2242175642fde481ea427f5bbb4c78ea0fd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-09 10:00:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/5"}}},"commit":{"parents":[{"commit":"c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7","subject":"Merge \"Vitrage: build images for Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-09 10:00:17.000000000","tz":0},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- empty one\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/517fd2242175642fde481ea427f5bbb4c78ea0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/517fd2242175642fde481ea427f5bbb4c78ea0fd"}]},"branch":"refs/heads/master"},"23c480d65136262fe5913d4dec6311274d3051d0":{"kind":"REWORK","_number":6,"created":"2019-12-09 15:17:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/6"}}},"commit":{"parents":[{"commit":"881edf3793deebc5bc57239b17d20f3dc04205d5","subject":"Merge \"Fix keystone bootstrap error message display\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/881edf3793deebc5bc57239b17d20f3dc04205d5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-09 15:16:50.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- empty one\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23c480d65136262fe5913d4dec6311274d3051d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23c480d65136262fe5913d4dec6311274d3051d0"}]},"branch":"refs/heads/master"},"2498a9da3cdde8cb742f948cb3d82318b5ae11b1":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-12-09 15:18:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/7"}}},"commit":{"parents":[{"commit":"881edf3793deebc5bc57239b17d20f3dc04205d5","subject":"Merge \"Fix keystone bootstrap error message display\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/881edf3793deebc5bc57239b17d20f3dc04205d5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-09 15:18:16.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single and multiple\n- Debian single and multiple\n- TripleO\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2498a9da3cdde8cb742f948cb3d82318b5ae11b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2498a9da3cdde8cb742f948cb3d82318b5ae11b1"}]},"branch":"refs/heads/master"},"cfffe50f5053eab5a5072a11d2957165a4af9275":{"kind":"REWORK","_number":8,"created":"2019-12-09 15:43:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/8"}}},"commit":{"parents":[{"commit":"881edf3793deebc5bc57239b17d20f3dc04205d5","subject":"Merge \"Fix keystone bootstrap error message display\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/881edf3793deebc5bc57239b17d20f3dc04205d5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-09 15:42:59.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single, multiple and one with missing repo\n- Debian single, multiple and one with missing repo\n- TripleO\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfffe50f5053eab5a5072a11d2957165a4af9275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfffe50f5053eab5a5072a11d2957165a4af9275"}]},"branch":"refs/heads/master"},"4ef52784aa73ffdfaebf34ccb171b1dac8105350":{"kind":"REWORK","_number":9,"created":"2019-12-17 08:57:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/9"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 08:57:12.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single, multiple and one with missing repo\n- Debian single, multiple and one with missing repo\n- TripleO\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ef52784aa73ffdfaebf34ccb171b1dac8105350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ef52784aa73ffdfaebf34ccb171b1dac8105350"}]},"branch":"refs/heads/master"},"5e9ed2233d81c6985e0996e7755373b8d5ef8de0":{"kind":"REWORK","_number":10,"created":"2019-12-17 11:57:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/10"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 11:57:27.000000000","tz":60},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single, multiple and one with missing repo\n- Debian single, multiple and one with missing repo\n- TripleO\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e9ed2233d81c6985e0996e7755373b8d5ef8de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e9ed2233d81c6985e0996e7755373b8d5ef8de0"}]},"branch":"refs/heads/master"},"540b0cf6d7da4f7a85c54af847d4f519471d6e27":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-12-17 19:11:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/697354/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/697354/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/697354/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/697354/11"}}},"commit":{"parents":[{"commit":"308e85aa413da84ed107b27fa09a3f0e0a64392f","subject":"Fix monasca-grafana being unbuildable due to old npm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/308e85aa413da84ed107b27fa09a3f0e0a64392f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-04 19:44:15.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 19:11:42.000000000","tz":0},"subject":"add more tests for enable_repos template method","message":"add more tests for enable_repos template method\n\n- CentOS single, multiple and one with missing repo\n- Debian single, multiple and one with missing repo\n- TripleO\n\nChange-Id: Ide4868d031bd84b1a6935aada7d0e7627f11bf37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/540b0cf6d7da4f7a85c54af847d4f519471d6e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/540b0cf6d7da4f7a85c54af847d4f519471d6e27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
