)]}'
{"id":"openstack%2Fkolla~697680","triplet_id":"openstack%2Fkolla~master~I5aedf1141862d51c5bb676a7393fc131c452c914","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-python-3","hashtags":[],"change_id":"I5aedf1141862d51c5bb676a7393fc131c452c914","subject":"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"","status":"MERGED","created":"2019-12-06 15:43:46.000000000","updated":"2019-12-10 13:27:20.000000000","submitted":"2019-12-10 13:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697680-1575984228875-1e73c108","meta_rev_id":"0f3b2217f7188b23863d2604b3f0cd3d53b70963","_number":697680,"virtual_id_number":697680,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-12-10 09:42:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 13:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-10 10:20:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-12-09 16:02:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-06 17:19:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-12-09 16:02:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 17:19:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-09 16:02:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-10 10:20:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-10 13:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94399a27c61ab5453da949ff367c864c0b77a0d3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 15:43:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522abdc7a6a5c9c64872359b95f84ab07f822c38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-06 16:17:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/80/697680/1/openstack-check/tripleo-build-containers-rhel-8-master/cb8a17b/ : FAILURE in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5d2f8d7445ac15dd0f2331abf9e5198504bd7ecd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-06 17:19:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26ff0bbf3a3a12ee75311a40bc206dbc837e2193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:38:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2dd88edddd9e4940b131d7931a0f8128 : SUCCESS in 29m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2534c1ed2e4940639addd850f05dad8b : SUCCESS in 42m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c8b5221ae07c4c6ab9335602cd8202c1 : SUCCESS in 1h 30m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1d9f19f459614cafa228b994ba73120b : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d82965cee88843529e8fb37894f2fcb0 : SUCCESS in 1h 49m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ae165320dcfb4eaf8206f587947d375a : SUCCESS in 1h 04m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d1e9cf6d3cb44032b916fc05fae462a2 : SUCCESS in 25m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d17e8d0baae49c1ab7c97de4f6ee4c7 : FAILURE in 24m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/338f401ffdd843229f219875f878f2d0 : SUCCESS in 1h 01m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d11a79bc66a4f12ab0e219d5c97525b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b126280353eb4ef39807bc8bf5508c39 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9884a8759db04284b101fbf782b1514d : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/502a37b2dac94c509f1268832f8d0ff0 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fcb5a21c0954a8d80c0e8268a8be534 : SUCCESS in 7m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3433370a3b04af8a0c6f56fef30c072 : FAILURE in 30m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df3fc634e4ff44cb91ea5c5b98f258a2 : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/659e87024e37462b9197d657c376c73d : SUCCESS in 1h 03m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6a40faa8d074bb9a60e475c54f71695 : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e67cdbbcfd5a4c04bb67ec4ee5cfa05b : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9f6d6ec19bd3c0ad34e9c4573b0a51eb65e4a830","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 16:02:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84bba753eb9406ccc5b77664e0f2bce08092db21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:02:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8368d63a530576da6b6d8de42886551f426bfdd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 19:13:01.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"852589f3b8d181a491233d634df955330c65725f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 21:29:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b4d83112029d4f3ea84d6f1bda25de81 : SUCCESS in 43m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/00a3351bbfe7409baaf8962e9d324e49 : TIMED_OUT in 2h 02m 32s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e0f5367a52c5410db46c7dc2d02a5735 : SUCCESS in 1h 31m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c34a762a35234c12828a6ac2aeed1a9c : SUCCESS in 1h 16m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/518933a36d40405c87368ab824e979c3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2452c622344e87b5a2bb9722aee4af : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d36eb994970c488e89f3e17527465bbf : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2001a7f945e4bb8ba6fe011699d7435 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64622dfe3ad4dd897ca46223e207d69 : SUCCESS in 7m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8940a4d3d8664fa8b2e30e261fe74626 : SUCCESS in 29m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3490deaf26564d6695de563646730c9b : SUCCESS in 1h 09m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3afe771205b488b9809383d88b77b62 : FAILURE in 1h 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c910160fea77608feefc40d328da4d3b1c5025d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-10 09:42:41.000000000","message":"Patch Set 1:\n\nrecheck\n\nnetwork weirdness","accounts_in_message":[],"_revision_number":1},{"id":"d5be8372cb9ff52f3f2e1e74191e5cc8bc19a588","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-10 10:20:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/80/697680/1/openstack-check/tripleo-build-containers-rhel-8-master/c28c53d/ : FAILURE in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c9510d2ea7fe2807091abad36d5795927c016980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 11:32:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f7c25a4b663349be8bbcf350a56cad58 : SUCCESS in 40m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3dd4d1adeae44c449673bedc5a339608 : SUCCESS in 39m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7548b4781b994dd3acb050b2aaef945d : SUCCESS in 1h 44m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7a5764c9b7194606b11364c2eadeb1ed : SUCCESS in 1h 25m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9a4a6672e42c4f2cbc66788b8e792bd9 : SUCCESS in 1h 34m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/39c43664ae944b2fb84d12046f7188b2 : SUCCESS in 53m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/68278249d7764a98b2a9de5c1ede7747 : SUCCESS in 27m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ac43519985d478eac6fc248da3765a2 : SUCCESS in 17m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77fc0a794b1c450b92a15cbacd5abcb5 : SUCCESS in 50m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/943fdf0fb7ca4b20b58827ef8853080e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4425186477e145d4b69fbe45e83e1a98 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d12fe2151d914654879b1dcd9cc8b44a : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfcfc12ab0d64b9a9fd4e211fc88f9fc : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6b5b7a2ea64f7aba28eafb527f8caa : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8378000ceec041e7b82d7c21bdda68f5 : SUCCESS in 23m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/511056ed49ed44d69a0c1c1281dda44e : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1cf0dfdb6cd54b75a0e9f915bea603a7 : SUCCESS in 52m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6bce1f1f0994c48a75a230007a49293 : SUCCESS in 52m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf3490bf5dab4977914721172aab476d : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4eeb1744cb08ba0a3b19c5588d60ad24206bf813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 11:32:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c502c462815ab1b938e218a9cdb6d7e0b3f17ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:23:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/010f4d667f4f49208aa0aed505439b26 : SUCCESS in 40m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/90b26e8e0fe04e15ba0fe4d271d26c0c : SUCCESS in 1h 26m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c9aa9bf6bf224b60bf98c3e20ffb13ce : SUCCESS in 1h 35m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f4cfbec642c54aa9958ce5f7c4c984a1 : SUCCESS in 57m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78213ca3d4394f49908ae640a05695bd : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed4a41d0b7a46cc90e88fe20c5b5e95 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09d8dd1a38c4156b65927efe37ec652 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5a9f8d1105f4d6db199fb96df35f4b4 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b280b5cec54431860e6b20d8619df9 : SUCCESS in 7m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c27cc318ea0e4107b43463ef3e933fee : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e7aee29312f4a988eba72c20ca8c0fc : SUCCESS in 55m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58adc619a3964349834de5656583a0dc : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2db2010858e472872cb4fc99d7d2990671fa28be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:23:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1af893b970d69765e28bf9474a5a644257abe5cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:27:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a077b8a0979341bc8c6168ca86df5e38 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ff14f85e46d789b53ef71c7aef628580e4cf334","revisions":{"0ff14f85e46d789b53ef71c7aef628580e4cf334":{"kind":"REWORK","_number":1,"created":"2019-12-06 15:43:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/697680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/697680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/697680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/697680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/697680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/697680/1"}}},"commit":{"parents":[{"commit":"55d8828be3d3476d6f3201682796fc3883ff615b","subject":"Switch to stable/train for source images on CentOS/RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55d8828be3d3476d6f3201682796fc3883ff615b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-06 15:36:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-06 15:41:32.000000000","tz":0},"subject":"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"","message":"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\n\nThis reverts the following commits:\ndf987c9d3fca40e0a7529e48a3626dfdd58e2168\nd866c63a32a494b8643b166c3b1e36e463a7502f\n6b40b655f124d0a95f1b7d89ce04236606a02010\n\nThese are no longer necessary as we are using stable/train branch for\nsource images until CentOS 8 / python 3 support is available and\nworking.\n\nChange-Id: I5aedf1141862d51c5bb676a7393fc131c452c914\nRelated: blueprint centos-rhel-python-3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ff14f85e46d789b53ef71c7aef628580e4cf334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ff14f85e46d789b53ef71c7aef628580e4cf334"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
