)]}'
{"id":"openstack%2Fkolla~697831","triplet_id":"openstack%2Fkolla~master~I85a505ca028878cb748e850613d07f89e240f2fc","project":"openstack/kolla","branch":"master","topic":"bug/1810979","hashtags":[],"change_id":"I85a505ca028878cb748e850613d07f89e240f2fc","subject":"Skip existing unmatched parent","status":"ABANDONED","created":"2019-12-07 10:51:10.000000000","updated":"2021-04-02 19:13:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03ab887dbaf3b226bf3ba4a1ce1811ad30223c96","_number":697831,"virtual_id_number":697831,"owner":{"_account_id":23831,"name":"Dawei Si","email":"dawei.si@qq.com","username":"davyyy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2021-03-25 19:43:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-07 12:18:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-07 11:21:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-07 11:21:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-07 12:18:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-16 13:02:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3b698e874b9b25cfdfcf59f0f2b6629aca7bb055","author":{"_account_id":23831,"name":"Dawei Si","email":"dawei.si@qq.com","username":"davyyy"},"date":"2019-12-07 10:51:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7165abc1afa04f2e3b3363bb995817404530dd85","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-07 11:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/31/697831/1/openstack-check/tripleo-build-containers-rhel-8-master/ef44064/ : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"34e7e443a9b1e78dfcf73375bd09bac02d198d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 12:18:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a87870573ff8427fb10743a0d675a1a4 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2b11ba5487b3436481b7287dacb2d217 : SUCCESS in 37m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3c5343a60f504fb8bb31e309aec49fcc : SUCCESS in 1h 22m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5026f157f0d04517b63ab906320d1da5 : SUCCESS in 1h 05m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/890d1e818b1a4c9eb45c38f757caa0c0 : SUCCESS in 1h 23m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fcb24ee5361d4d57b3c9dbe5e05c699a : SUCCESS in 48m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e034e1fa851a4bc391c67779db2d27e9 : SUCCESS in 23m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/19fb9574113f447885535f01edd6e633 : SUCCESS in 19m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a86c016bdc14d9cacddb45010b32791 : SUCCESS in 59m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e571d61a1bac44d8a259168ec9140ec9 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a8372dedbd41b09ebc4f0143846882 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55f351c9f6ad43d596ddbbd944e2df7a : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e98d6b8bbb74a44875b51a45d8a8e76 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b96ce626d84b6eaf5249acf6d32708 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0cb4540e63e6498e86b453e9cf5f7d8e : SUCCESS in 41m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd1954a4a11c49b7ad6a30e1645120b2 : SUCCESS in 42m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c387efaa4ce84a9ba8b83f49ca4b83fd : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c00ff6cda40c420684cb9e05974f7e87 : SUCCESS in 55m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f527d663523e474fbe5fb75e698d1689 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"225ddec6ad1dfcb45232b6463357fa307e074433","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-16 13:02:16.000000000","message":"Patch Set 1:\n\nThanks for your patch. I have an alternative proposal which I think fixes some other issues. Could you check if it fixes your issue?\n\nhttps://review.opendev.org/#/c/713221","accounts_in_message":[],"_revision_number":1},{"id":"95413aadaaf812fe3d326877a31646e7a425f089","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 19:43:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thanks for your patch. I have an alternative proposal which I think fixes some other issues. Could you check if it fixes your issue?\n\u003e \n\u003e https://review.opendev.org/#/c/713221\n\nThat patch has now merged.","accounts_in_message":[],"_revision_number":1},{"id":"03ab887dbaf3b226bf3ba4a1ce1811ad30223c96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-02 19:13:59.000000000","message":"Abandoned\n\ncleaning","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b87b12e045c036b3f7c331a13e01a5040a5e287","revisions":{"1b87b12e045c036b3f7c331a13e01a5040a5e287":{"kind":"REWORK","_number":1,"created":"2019-12-07 10:51:10.000000000","uploader":{"_account_id":23831,"name":"Dawei Si","email":"dawei.si@qq.com","username":"davyyy"},"ref":"refs/changes/31/697831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/697831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/697831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/697831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/697831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/697831/1"}}},"commit":{"parents":[{"commit":"c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7","subject":"Merge \"Vitrage: build images for Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7"}]}],"author":{"name":"davyyy","email":"dawei.si@qq.com","date":"2019-12-07 10:32:55.000000000","tz":480},"committer":{"name":"davyyy","email":"dawei.si@qq.com","date":"2019-12-07 10:32:55.000000000","tz":480},"subject":"Skip existing unmatched parent","message":"Skip existing unmatched parent\n\nSee bug: #1810979\nReleated Change: I6b6897d20626825bcd235b9a85859644effbd222\n\nThe change is fine, but when the parent is skipped, and unmatched,\nwhich is a common case, filter_images change all other status to unmatched.\nwe should keep the skipped status.\n\nChange-Id: I85a505ca028878cb748e850613d07f89e240f2fc\nCloses-Bug: #1810979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b87b12e045c036b3f7c331a13e01a5040a5e287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b87b12e045c036b3f7c331a13e01a5040a5e287"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
