)]}'
{"id":"openstack%2Fkolla~697874","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I74142928d97af932a04e33476771360579e01ee7","project":"openstack/kolla","branch":"stable/train","topic":"bug/1853474-stable/train","hashtags":[],"change_id":"I74142928d97af932a04e33476771360579e01ee7","subject":"Ensure lsb-release exists on debian/ubuntu","status":"MERGED","created":"2019-12-08 09:06:03.000000000","updated":"2019-12-09 17:47:34.000000000","submitted":"2019-12-09 17:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697874-1575913557386-19300d8e","meta_rev_id":"1e0a519370fda231bb6028942a8216e2088973a4","_number":697874,"virtual_id_number":697874,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 17:45:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-12-09 15:59:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-09 09:49:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-12-09 15:59:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 09:49:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-09 15:59:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-09 17:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93ee5b863a6e6450c5b062e59dca67b100203736","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-12-08 09:06:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"89af02384bf61893a53ca27318dd849a925d9964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 11:20:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e24d6ae5791044459d35defabdd99984 : SUCCESS in 51m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/65a06ce0c6994815b33fc8d636390a3d : SUCCESS in 59m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a123346b501d4e448d49bbcce21c5fec : SUCCESS in 1h 10m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2950b404fae6488aa3e84a186292528d : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/49b78cad5fb040dc8b2efa1c61610bbe : SUCCESS in 1h 19m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/33fbb8409c584b4fbccb820dc5b85d4f : SUCCESS in 57m 25s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/6ba030b1047c44e18a1b8152f0b06a47 : SUCCESS in 41m 24s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/77d7227ca6bf43e6bdaf9b4bb40e296b : SUCCESS in 2h 01m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8b443d120e5a420380f601979a076842 : SUCCESS in 2h 05m 23s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/5cd78340fc694346a8f407d728bdf0e3 : SUCCESS in 1h 45m 32s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/513ed80966544683aa9b6df680671379 : SUCCESS in 2h 00m 07s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4681e410a2d24685aee3df9bcdaf50fb : SUCCESS in 1h 30m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3207778ba43464d87121e281d782f75 : SUCCESS in 22m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e205478c79a47be827de070b9c7f579 : SUCCESS in 16m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eeb123f38f4447539a7f2664d68f1572 : SUCCESS in 1h 01m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae1e3a9d243844bda18a5542c43dbc79 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6967311bec439d8515a9dac890cf82 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7857076f1666477c99e16f0b82119b2c : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0122923e2ac41b289db7a8354e571e1 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bac72b60d8f4b07ada8facac639c926 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63c56d9e8254f528ed2a824135a70a1 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d98a61e1d5a347dabd10dfe444884889 : SUCCESS in 25m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f91edeb9f37426a8acb768b5606d512 : SUCCESS in 27m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5074b2385954cbb830ec45728fb4083 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65256202c99e4a03b7bfac631c887a68 : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e8fd5abd5ac451daccadf19d309f9fd : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d74a2252c34e47749cd827032a8a6a68 : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2a0a798d3f48447faddb4484564ed1ef : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a160e355c43c42dbbec3ffa805cfc1ea : SUCCESS in 2h 05m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/61c7a84eed074149aa737b999feef31d : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cf7d42d40a074681b477e8a972d9b6ce : SUCCESS in 56m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b184282cd6e3bcba99c722cb297d1ed6045f59","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 09:49:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5c268061b65ccf4fb27dd991d70f02ad28840bc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:59:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"336e217e622b89e30fb92588e66ea34db218d131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16102e359e7b9aaed0934774c7fcdf205932d9d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:45:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab559d7b64fec1c19032eb122d4a9e0553a7c72f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:45:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7d457ea581c44c6dbbed2735566d9a4c : SUCCESS in 48m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c7c6169ea53c475aaace4ee8bea7f508 : SUCCESS in 1h 27m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ebfce24589454d40ac68e3d64766ea37 : SUCCESS in 1h 30m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b9776e4f56bb48dc87234f3ee60be098 : SUCCESS in 1h 00m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af33ea7acf514c5cbfb7e3026743fe9e : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6dd2f4453e4a2ea647a38b60af7b70 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a943617f60a4af7a583c1511b293d9d : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f518906960d419a9af96fc62e480e2c : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2044534142da4212be3701ee86292c13 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5a75d3a2814f1f92f3f953a114e438 : SUCCESS in 7m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e4f0708171c41598b8df77594ac991e : SUCCESS in 38m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92fd06ba3fad4b11b2e152cd6de3a9e9 : SUCCESS in 1h 00m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b958eaf9c7847dc80ee301d1439a4e5 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7ab94efbd334f1e006e990a8306e7abe071b0625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa3492056a44c8187a60bc81c22cca4 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8c743092ca873543b44794489c62e9c8d2e2d80","revisions":{"b8c743092ca873543b44794489c62e9c8d2e2d80":{"kind":"REWORK","_number":1,"created":"2019-12-08 09:06:03.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/74/697874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/697874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/697874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/697874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/697874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/697874/1"}}},"commit":{"parents":[{"commit":"19c38d9c9d6416b7859037396484fff2e587ec4b","subject":"Install etcd3gw to fix Ubuntu binary tooz coordination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c38d9c9d6416b7859037396484fff2e587ec4b"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-22 07:36:38.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-12-08 09:06:03.000000000","tz":0},"subject":"Ensure lsb-release exists on debian/ubuntu","message":"Ensure lsb-release exists on debian/ubuntu\n\nChange-Id: I74142928d97af932a04e33476771360579e01ee7\nCloses-Bug: #1853474\n(cherry picked from commit 05c84fdf86d62545043cd36b984ca7627308d63d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8c743092ca873543b44794489c62e9c8d2e2d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8c743092ca873543b44794489c62e9c8d2e2d80"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
