)]}'
{"id":"openstack%2Fkolla~697982","triplet_id":"openstack%2Fkolla~master~I82444bc2272311023cc9e92c5a298d1c4c87483b","project":"openstack/kolla","branch":"master","topic":"bug/1855701","hashtags":[],"change_id":"I82444bc2272311023cc9e92c5a298d1c4c87483b","subject":"Fix keystone bootstrap error message display","status":"MERGED","created":"2019-12-09 11:13:33.000000000","updated":"2019-12-09 14:21:09.000000000","submitted":"2019-12-09 14:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697982-1575900715656-8628b750","meta_rev_id":"0d0c9b2460c9f9f29e478646dddf1e1b81e4b7e0","_number":697982,"virtual_id_number":697982,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 14:11:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-09 11:52:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":2,"date":"2019-12-09 11:37:58.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2019-12-09 12:18:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2019-12-09 12:18:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 11:37:58.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2019-12-09 11:52:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-09 12:18:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-09 14:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28cef114b593975de2eac5fe57567da39c4dd9fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 11:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1af0bac7a90cbbc3e3393e18f974a5791b47f815","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 11:14:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5e2700a0bca3751a29c06b40afdf51212b9affe0","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2019-12-09 11:37:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7c12a3b37e8cf22f8161f19e7d3b05e4a2ec28c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-09 11:52:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/82/697982/2/openstack-check/tripleo-build-containers-rhel-8-master/a917129/ : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ea585f8ffe873a9ce7cf5b52ef90849ab4a5d5d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-09 12:18:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3caef26ca574effc82ca7764c6b2279e6605c229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:43:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/158970a40ed342959cc048307a2c9f18 : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/51f9e81d08ea4a7382ca5513bb3bd733 : SUCCESS in 1h 06m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ad646c6a6aea43fc977a0a99749c052d : SUCCESS in 1h 21m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/277f7d8387eb44919254a09cdd2890b1 : SUCCESS in 1h 05m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ac15e853d9964797abd65b778209d416 : SUCCESS in 1h 27m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2e7b390509d24e60adc4ef7047ed5740 : SUCCESS in 1h 01m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bf10e7a01c0489eb8750333b581da1a : SUCCESS in 22m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/06614eb2c37a4bf580322b946d1a5426 : SUCCESS in 33m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/597e4bcfbf564027bd7c950a95ca817d : SUCCESS in 1h 08m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef74aa0b274440b19c9a01f3ccc01e50 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb518dbf0c64c8986b0293e424adc8a : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c3d32cb773f4c7daf10902013fd755f : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08a42d772b7c4cba885ef76d6930a19c : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9431e6eedf744068b32a3b6c8a86c3d5 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1ec5984dd3f451ea563065dc52b178d : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66a15eb2b886492382381ee8dae4b77f : SUCCESS in 37m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92fc5778e03d4853b6cdd2a3be00316d : SUCCESS in 34m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/872a671e78a948fabeabbf1220f67bca : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce4fff5822134bb58a754f8c9efc45f3 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/910741b6dede4cedaa350a7353fd5a37 : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"666a155eeb16459313b96f4185f1d0ccb09ddc1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:43:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53ec8dab05e5185f527c9f00753c85c259ba53b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 14:11:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"04b3c8f586aee265d688b44d7c5d63d49c306e6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 14:11:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/43bae04e5bcc47d8bdad01c83a58d8c9 : SUCCESS in 35m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7cd7745c647048cdb702aa1ff3d71007 : SUCCESS in 1h 24m 02s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dbc9d4c2434d485fac19a2ed84ceb370 : SUCCESS in 1h 27m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d61dd9ada75e40918a5863b71143da09 : SUCCESS in 47m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1cda11cc9134b26ad8e82f031609bc1 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f70812ddc743a8b491236a2908264d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f143dc12cc642f69ea6915111e0bdbd : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d413ed866857487196753751780410ae : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d186922dfc54489da06bb2847ce557e1 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53d85464ee24f38813990031cd7b2ea : SUCCESS in 4m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6808ca20ad8d4ce28a4197d85ff3f653 : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/298cefcf9a734636ae9996c6e3d48b68 : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c40e4d3df234f30a51740b8a176a0ad : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d880ab88ec050e80f5018e6fd31568254162a6ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 14:14:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8573e993e74f2abc41d7520c42d520 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c35918654724bd4b713c157bf2170dd : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4dc46d2e14f6628ffd6509f46a89c756f4323149","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 14:20:53.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 17b16cde7a38f56052f4397fbd707c33b2af0555","accounts_in_message":[],"_revision_number":2},{"id":"93b56c3e39620b00db591175370ad703c640c744","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 14:21:01.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54","accounts_in_message":[],"_revision_number":2},{"id":"c9d5ffb9636068288a453445a17d585d6cf4ef4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 14:21:09.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit aaf34cfa2570a65fb36fea12f38edeeb3ad2a4db","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcca3fabe305c1e206fdb231c31809ceff2d4774","revisions":{"39ffbbe9c1669bdafb8417c26672ed02451b3823":{"kind":"REWORK","_number":1,"created":"2019-12-09 11:13:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/697982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/697982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/697982/1"}}},"commit":{"parents":[{"commit":"c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7","subject":"Merge \"Vitrage: build images for Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 11:11:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 11:11:26.000000000","tz":0},"subject":"Fix keystone bootstrap error message display","message":"Fix keystone bootstrap error message display\n\nIn some situations, Keystone bootstrap can fail, and then unhelpfully\nfails displaying the error message output by the \u0027keystone-manage\nbootstrap\u0027 command. This appears to be due to unprintable control\ncharacters in the error message which prevent the output of the script\nfrom being valid JSON.\n\nThis change fixes the issue by piping the output through \u0027cat -v\u0027, which\nreplaces unprintable characters with control codes.\n\nChange-Id: I82444bc2272311023cc9e92c5a298d1c4c87483b\nCloses-Bug: #1855701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39ffbbe9c1669bdafb8417c26672ed02451b3823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39ffbbe9c1669bdafb8417c26672ed02451b3823"}]},"branch":"refs/heads/master"},"bcca3fabe305c1e206fdb231c31809ceff2d4774":{"kind":"REWORK","_number":2,"created":"2019-12-09 11:14:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/697982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/697982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/697982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/697982/2"}}},"commit":{"parents":[{"commit":"c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7","subject":"Merge \"Vitrage: build images for Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5cbea5d2932d68bbf572c6c856aed8a7fe9b0a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 11:11:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 11:14:36.000000000","tz":0},"subject":"Fix keystone bootstrap error message display","message":"Fix keystone bootstrap error message display\n\nIn some situations, Keystone bootstrap can fail, and then unhelpfully\nfails displaying the error message output by the \u0027keystone-manage\nbootstrap\u0027 command. This appears to be due to unprintable control\ncharacters in the error message which prevent the output of the script\nfrom being valid JSON.\n\nThis change fixes the issue by piping the output through \u0027cat -v\u0027, which\nreplaces unprintable characters with control codes.\n\nChange-Id: I82444bc2272311023cc9e92c5a298d1c4c87483b\nCloses-Bug: #1855701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcca3fabe305c1e206fdb231c31809ceff2d4774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcca3fabe305c1e206fdb231c31809ceff2d4774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
