)]}'
{"id":"openstack%2Fkolla~698020","triplet_id":"openstack%2Fkolla~stable%2Fstein~I82444bc2272311023cc9e92c5a298d1c4c87483b","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1855701-stable/stein","hashtags":[],"change_id":"I82444bc2272311023cc9e92c5a298d1c4c87483b","subject":"Fix keystone bootstrap error message display","status":"MERGED","created":"2019-12-09 14:21:01.000000000","updated":"2019-12-10 13:25:21.000000000","submitted":"2019-12-10 13:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698020-1575984230335-7edb6769","meta_rev_id":"42e7eedd286ac24344795d79dea0b6e48b83cc5a","_number":698020,"virtual_id_number":698020,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-12-10 09:40:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 13:23:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-12-09 15:59:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-09 14:23:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-12-09 15:59:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 14:23:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-09 15:59:54.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-10 13:23:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da561a2fd75c49d057a1cefd33db79388ad6aaa8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 14:21:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6fa80017b380ae58f2e621868a9905afe7594b46","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-09 14:23:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ba5c0aa254b4c681b10af19a82241a41676ab1c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:59:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44a011dd82b90ef297dc4082402584596967afc2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 15:59:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"76d18ed0f49b9569042a24782fefafdf9b1036bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ded3415bdc814e25a19205eeb99e03d8 : SUCCESS in 40m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/baf67299721c4eb5a4757d1dba8fc28c : SUCCESS in 45m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a089575ebcc345028b03894cfe043b36 : SUCCESS in 1h 24m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d94cb489009b48099e882127bcd111e7 : SUCCESS in 1h 09m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d1157aa29be7458d955d0effba42f70e : SUCCESS in 1h 36m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dc5dec63635e4b59b3e9995ef6fb0354 : SUCCESS in 58m 54s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/52a54d19729c4534aa6b332c981d0f36 : FAILURE in 38m 35s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ee8bc4a6750842b982c7c576cbea0ea8 : SUCCESS in 49m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aea19d699b7843a4b279e36a7e161dd1 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9aca8b55a334f4f820edd464d99daa4 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea66d3d711c8493f968313a7f3898e9b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b626931e40b4c2baa1e2c8607b92b87 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/254812ee64134956b006bf21d7e999f3 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0affb82bee544b3ca0163e55e9ab32c8 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f227e21b2514fc19be1d8dc3a66e123 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3348c187598e4b1891fcb818b326503e : SUCCESS in 32m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b22230f8eb3444ebfb0765209c13bcd : SUCCESS in 34m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/caa77a0e3f954dc1b554e23b78d89c92 : SUCCESS in 1h 09m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47d5e4aa51c04617805390fa8223b2ef : SUCCESS in 52m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b998439bb25341c2a1f40af0729f7bcf : FAILURE in 39m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a68c4260455a65f9cf78a70e38328cc236849dcb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 16:33:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6691bf230b3e58fe91eb80d87b50f3c91e162918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:21:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a12ab09ce3fb4fbfb44f217e0389cac4 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b8f660b197424bff92314eae50ca7077 : SUCCESS in 1h 10m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4372138ca8db4914a6ec2010cdc460ef : SUCCESS in 1h 09m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/489e4f0c8df0411985c7f8b737b52ded : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e9372fd4c524bdb854d7f579b06c3d3 : SUCCESS in 1h 43m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c5aceb1421fa4b39afcf95140d7a2b17 : SUCCESS in 51m 56s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c96e9e7e49524934be914698e2c3c06d : FAILURE in 47m 01s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d5a4319a649042b191d691e7b5fedb5d : SUCCESS in 47m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f88dd6a2f504551a76f94e502a3e8c6 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343e33690aee448a94d419a960c0a34e : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d4dc2fa9567462da69658fadf526917 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b66f6953b6046ef8c17852547fd37a2 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf236fb487b846d5802ed119783fc45d : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc67b7b6d3c94e5b9eb4e6fe478cdc9a : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8f7fa74d3154ed4a903aa729ae4ca4f : SUCCESS in 4m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ac11069608049619a825494689048ad : SUCCESS in 24m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2d7fae9590f4f12991dae79ce21ac7b : SUCCESS in 29m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8a659a8dae0401eac4bb5b271e1880d : SUCCESS in 51m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ad7d745ea344bcc9ddabdf3b9125d2f : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30a0d320ecd348aa8a34e39ea44a873b : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ed62910e97bfe5715ef532f3e23747a4d0412ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:22:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f349d3554eece83c48267bbfdc5b8842b2b17bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 19:13:20.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f36e1e0d2d06632160aba97669c99f3cdd6030fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 22:33:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4b18333220ca4b19a459eba4e028ce57 : SUCCESS in 54m 01s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f75b23d1302243769f5861fc7a198981 : SUCCESS in 48m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/affedb5b515641548532c9076d362e43 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8b090f45be4aca86b14f9cfdbe8f37 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e64a8615348845bea8f49342f4db711e : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9ae29adc42409ca6004517cc09956d : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cc7861b48e44af5aaa99e7de064d542 : POST_FAILURE in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dedac4a930543cbb717cb7b3f34631b : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/398689bece244b6095239a0721da9896 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93d6b7dd150a43419cb1a47afc493fe0 : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3ea985c4031048474c8bc54ea58a1277b8b338b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-10 09:40:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdc96c371120d1b258507cf1410b6433dab82535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 11:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ce545274f93340f9a9454f24c0a1e76b : SUCCESS in 49m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/62e9d02679a64a69b3c5f1e5d8cdbddd : SUCCESS in 52m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c48fe67f5142407c867e23a1a23755b6 : SUCCESS in 1h 18m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/af0e00506c0442ffa1899cf8f317b4bf : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7b918a17bdee4382bf134823d47f3d17 : SUCCESS in 1h 51m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c08f1e90a29e4e328d4a36be4149fe30 : SUCCESS in 54m 52s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bf9b99de6c3946e0a5ed3ee40fd28dea : FAILURE in 48m 07s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/640c1121768746e38e0822ee0b8d26c7 : SUCCESS in 51m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28acd22bd00f419ea70fd46c0ee3c417 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b368829284db4372af67afea59764df9 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5328fd39b6504580af994e1fe9111fb1 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab93891aa42a468eaa18b026d9f6a4d0 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10415b3ebf95490b8a5fb54eef2eaab3 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcafbf112c74937ba928395cdd82ff2 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/251691f62fe54b318105435134bf9e2c : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cbb836adb014504ae65129fde54e966 : SUCCESS in 27m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9df4ce19d027459da6acacb0a38088ca : SUCCESS in 33m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e859357a6be74a83867729e116036a27 : SUCCESS in 58m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5349ef599a304a459ef6b2a646acbfe9 : SUCCESS in 49m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9a75e8a354443bdb4a9151308e74e4e : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9e0eb882c42dba3f4f6e6b20b00797b633e53649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 11:34:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f86ae0d8007ec2ccbc973fd32f5a2df571cf9947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:23:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1c2685f35a0b41f48a64fb969f4b2cd5 : SUCCESS in 57m 42s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2d969fa5dc6547e296b559bdde3bdf7d : SUCCESS in 57m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b2a52d89a4244b7a40292ec9858741d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23ab08463ed42e3bd14507e13d55c1e : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8acc024a255d47d7919252d516c1aebc : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed1b644be8a7433297c4560344b4927f : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f12f2bed2754e8da467e22ad6e33753 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1cc214bd9934edb9f7e2a4820b1c4cc : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c9c3678eb340fd8f2fe36c9f7d03b7 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/948133ca2c594380a16be44c7099516b : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"82433efdab262e1177440c67bf5a506f9bde949b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:23:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e272a7bbad527f7e06817c97fc58f9a30960283","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:25:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7176a2662b924ed6a12ad7fcf0a3b2e8 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b5d05204c6d4dfd9e07a390c6c46722 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54","revisions":{"7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54":{"kind":"REWORK","_number":1,"created":"2019-12-09 14:21:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/698020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/698020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/698020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/698020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/698020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/698020/1"}}},"commit":{"parents":[{"commit":"38191924fe0400415a1d20a17529a9e1003a30de","subject":"Install etcd3gw to fix Ubuntu binary tooz coordination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38191924fe0400415a1d20a17529a9e1003a30de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 11:11:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-09 14:21:01.000000000","tz":0},"subject":"Fix keystone bootstrap error message display","message":"Fix keystone bootstrap error message display\n\nIn some situations, Keystone bootstrap can fail, and then unhelpfully\nfails displaying the error message output by the \u0027keystone-manage\nbootstrap\u0027 command. This appears to be due to unprintable control\ncharacters in the error message which prevent the output of the script\nfrom being valid JSON.\n\nThis change fixes the issue by piping the output through \u0027cat -v\u0027, which\nreplaces unprintable characters with control codes.\n\nChange-Id: I82444bc2272311023cc9e92c5a298d1c4c87483b\nCloses-Bug: #1855701\n(cherry picked from commit bcca3fabe305c1e206fdb231c31809ceff2d4774)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
