)]}'
{"id":"openstack%2Fkolla~698027","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I73b1964b7b22f1a2d5637209b86559b79602131b","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I73b1964b7b22f1a2d5637209b86559b79602131b","subject":"Vitrage: build images for Debian","status":"MERGED","created":"2019-12-09 15:11:07.000000000","updated":"2019-12-10 12:40:49.000000000","submitted":"2019-12-10 12:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698027-1575981536547-758459c9","meta_rev_id":"89281b011388bcfdcbd15f4c1dd48d0982f380f9","_number":698027,"virtual_id_number":698027,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 12:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-12-10 10:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-09 16:00:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-12-10 10:47:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 16:00:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-10 10:47:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-10 12:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"104f8d3ec232d56527cc7514145c3095c15d054e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-12-09 15:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c16689e9570fc134423da66c006775405e4788a6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-09 16:00:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"789ff41fda9ed8e93e0fcdc7d18834227ec1a175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:44:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0a6e033b7df84ac7bffcd925167abb90 : SUCCESS in 47m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4c86b9eaea314fa19e2ccd98a4f73ba1 : SUCCESS in 53m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a11fe3a719c841778698eda6521bf186 : SUCCESS in 1h 26m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e14fbb3e2fbe4288b1f33cad20de7792 : SUCCESS in 1h 19m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/972a1df73ea94517baa8cb7b04dff143 : SUCCESS in 1h 27m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9f6e4bf2392479fa1e90c0807b60d39 : SUCCESS in 58m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3285b114cf5543a5880016f85ed40274 : SUCCESS in 24m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e2f61365731c4beda499baa38bdf88dc : SUCCESS in 31m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/954778fac08a4f51a3bd4d0f9aff8e4e : SUCCESS in 59m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ce8c7076c5d4ed9b2bb202d119b10c0 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/222040e5325b4c81b49330ea8d63caff : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24e43f83949d45f18cf8acd75c866ddf : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7228450c644a61a9966c033614ee98 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51c58f58ddac4d80b6e5a1075b248323 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed3375521124e0aabad8c2a11009225 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d52f72be47647d58da73731bee5d4cf : SUCCESS in 24m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/318e07e2482247a990138c174311ab91 : SUCCESS in 36m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7cfe844ebed416ea538f0066cb8aed8 : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee779388cda14ee9b198bfe82de38f95 : SUCCESS in 54m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37ddbe5838e04e02871f83db965600b1 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6b12ebbdd15ec2923edf69489e307de4b294a9a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-10 10:47:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07f2c300c7f87c228894f15911bc108b8efe0e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:47:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3edf5c3d2db92acaaa31a65a40a00122c244b93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 12:38:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e3b3273842724e869e37c6356ee2cfca : SUCCESS in 1h 06m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6fffc17802b14c7cb5a1743dd0776f89 : SUCCESS in 1h 36m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bae2d233669647e2b2b7bf093dacc7db : SUCCESS in 1h 46m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/26cbd30ce00c49268ac67c240881bd50 : SUCCESS in 55m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16243022274046d59506a052f1ae3203 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27114640fb454ceaaba7889968c6dcd3 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/387be255919942a0b2c6c4dddd95a8be : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e71e6012988d4876b4379cbbc605804c : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71d37a0da79f42b08e2641f54c9102d2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb6c9eeb47c4dccb28f5051b265f468 : SUCCESS in 8m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0233bbabcfbe445687d98eec52382a93 : SUCCESS in 50m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca372ba2414c4fba88bd0c0c96cde499 : SUCCESS in 55m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aafe01f29bb94b82b8a6b92dc6f12da3 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"836224bd711816869823cb19d8ccb3243aff3291","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 12:38:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"89281b011388bcfdcbd15f4c1dd48d0982f380f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 12:40:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d2e3bee6fd43438b2b002794368323 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa70406628c4be203150ababc0f293fcc54910b0","revisions":{"aa70406628c4be203150ababc0f293fcc54910b0":{"kind":"REWORK","_number":1,"created":"2019-12-09 15:11:07.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/27/698027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/698027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/698027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/698027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/698027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/698027/1"}}},"commit":{"parents":[{"commit":"19c38d9c9d6416b7859037396484fff2e587ec4b","subject":"Install etcd3gw to fix Ubuntu binary tooz coordination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c38d9c9d6416b7859037396484fff2e587ec4b"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-11-06 22:22:01.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-12-09 15:07:29.000000000","tz":60},"subject":"Vitrage: build images for Debian","message":"Vitrage: build images for Debian\n\nChange-Id: I73b1964b7b22f1a2d5637209b86559b79602131b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa70406628c4be203150ababc0f293fcc54910b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa70406628c4be203150ababc0f293fcc54910b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
