)]}'
{"id":"openstack%2Fkolla~698210","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Idae69105739a57937fb5ad596078dafc6a32f1d5","project":"openstack/kolla","branch":"stable/train","topic":"bump_skydive-stable/train","hashtags":[],"change_id":"Idae69105739a57937fb5ad596078dafc6a32f1d5","subject":"Update skydive version to 0.26.0","status":"MERGED","created":"2019-12-10 10:35:33.000000000","updated":"2019-12-10 15:56:13.000000000","submitted":"2019-12-10 15:53:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698210-1575993206968-51380ceb","meta_rev_id":"1c6699102e03ccf02fdbf20dae6cc09018e25a85","_number":698210,"virtual_id_number":698210,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 15:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":2,"date":"2019-12-10 10:46:14.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2019-12-10 14:18:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2019-12-10 14:18:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 10:46:14.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2019-12-10 14:18:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-10 15:53:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f47a572ee0328fbdf14f4ae4f1fe14fafbeae3f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-10 10:35:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ab12c9a662e9e8b3cf1ef4ab0088c5567b2d9057","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2019-12-10 10:46:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66113c6a68d774a36878a2e40f8208188d8de6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 12:21:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9e73b2356c6b4811abe0e06690a49fc2 : SUCCESS in 54m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/39050684792745ceba2944415e24ffba : SUCCESS in 43m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3d94adec578c4989a51b1d61f93a6bc5 : SUCCESS in 1h 43m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6beff5d0ce144de39c1984dca17b548f : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/30aeab10e5b542c08a58c22de1945bdb : SUCCESS in 1h 40m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/160d8e03d979403ca325b836a1700894 : SUCCESS in 55m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cc556e50059b4ae081ada31504c962c0 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17fdf6e10032474096d1b1e1452bee5e : SUCCESS in 19m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa110e2b39c5478ebaa1783869ca07e1 : SUCCESS in 58m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa63ad9834754bcaa71de048a59dca1f : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6007774d8a442c8968bf8422bab0ba : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95fffe295b534c4691f243a53247829a : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c22f0f95cea741dc8ac6c5117effbeeb : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5b07b59d74f4858a8bacd6af1dfdc62 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e73e5910c14311a6c7876b30c10841 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed4d0dc2340f4264ad1aee221d4de7bb : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9ea97b65e3a401cac2e9bbff84578d2 : SUCCESS in 31m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b900ca6a17ad4b7a98ae6688d179cca0 : SUCCESS in 1h 02m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4741ccb3d22a4688963291c3524a0301 : SUCCESS in 52m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a936d930194446bebcd541da21e3bcec : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"76d4a6d5ce1da18a0cd9e16ba536f54b888e815c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-10 14:18:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d777517d4f3a77e0a3e207032e3e9f6fec8469b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 14:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73706651e44cdfa8b58babe032735a375bdfaefd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 15:53:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/51965f67c0424bad81affe95057d3fbc : SUCCESS in 1h 00m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/42776f09bfff4b27899e3a6b1474a105 : SUCCESS in 1h 25m 03s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/17ad436209be4996b2b429b3a87f590a : SUCCESS in 1h 21m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8f1b9961ea6d4ff3941b51e6e3df166f : SUCCESS in 49m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/926a10bdf79e4814a7476a477d73e438 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0435b4519884155ae1fe21b5fb07178 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53e84c6973924fd387a69da52d31c9a6 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c1d6cee0d7c48c59740f6914ee20cc7 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f5a78245b2d46a3b0f70619bde4b726 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6951c3fbae41c5ba1c185b53f4f31b : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a259b6709eb464abc753775c3a4f948 : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11a85448bb824114bb442cf59e6759e3 : SUCCESS in 1h 26m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8cf98e4d37974b2db9e3011e1cc3f9d7 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f8b625bd48eed8ce984983d1b363ea1346ab44c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 15:53:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bbe5ad42abb00500a8cef3ddacd2b82eb04f12a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 15:56:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a79feea6f54d36b0a4ca13a25d5efe : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e6d7c7a09113c80a2ae73f712265b1c7ac23034","revisions":{"5e6d7c7a09113c80a2ae73f712265b1c7ac23034":{"kind":"REWORK","_number":1,"created":"2019-12-10 10:35:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/698210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/698210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/698210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/698210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/698210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/698210/1"}}},"commit":{"parents":[{"commit":"5ef72d632881dfb07e622abe0da3304f6c25d83c","subject":"Merge \"Fix keystone bootstrap error message display\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ef72d632881dfb07e622abe0da3304f6c25d83c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-20 11:06:53.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-10 10:35:33.000000000","tz":0},"subject":"Update skydive version to 0.26.0","message":"Update skydive version to 0.26.0\n\nChange-Id: Idae69105739a57937fb5ad596078dafc6a32f1d5\n(cherry picked from commit c82ab3cdcec673eb4f0ed38f4604bf21301d804b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e6d7c7a09113c80a2ae73f712265b1c7ac23034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e6d7c7a09113c80a2ae73f712265b1c7ac23034"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
