)]}'
{"id":"openstack%2Fkolla~698430","triplet_id":"openstack%2Fkolla~master~I4778db57dae0953f5c7116105466c2830592b514","project":"openstack/kolla","branch":"master","topic":"ci/add-centos-aarch64-job","hashtags":[],"change_id":"I4778db57dae0953f5c7116105466c2830592b514","subject":"skydive: mark as unbuildable on !x86","status":"MERGED","created":"2019-12-11 09:33:27.000000000","updated":"2019-12-11 16:43:55.000000000","submitted":"2019-12-11 16:41:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698430-1576082515250-fee75b61","meta_rev_id":"1b83db43cb877be6d016701b3b39e8e06f1ddafb","_number":698430,"virtual_id_number":698430,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 16:41:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 13:41:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-11 14:56:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-11 13:08:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-12-11 14:56:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 10:04:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-11 13:08:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-11 13:41:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 14:56:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-11 16:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d076df93dc1cdc03be330c69108e9a982f8a54c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 09:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"962b2b9ebe99bfb09a19a0816cad9b8c8602865f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-11 10:04:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71962b06a2665c0582b72749629c9c0a11b30123","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-11 10:06:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a9103b31585141dd9cba921e1539adcf : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e00073c2b3574b91c83abbc19932ae13141fa347","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-11 10:31:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b6ddf804fa12e8f5aabb1eb9ec5953b4824f1e4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-11 10:49:10.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nneeds update of matrix, possibly releasenote too","accounts_in_message":[],"_revision_number":1},{"id":"5bfad99fe6bfb6ed29b24b2a2f22839c7247a3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 11:04:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b39e038919d44fbdb2d608567001afae : SUCCESS in 40m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fdfd84d104e04f0c835b437d01800211 : SUCCESS in 42m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/df2ac3b18ed746db823da8b76eb30232 : SUCCESS in 1h 25m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6fe8609996c346e2aeaea52f195ebef2 : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6d643069916b444aa0fbb88d32b57bb3 : SUCCESS in 1h 25m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4873cfc68fa541a199b51667f93f7d29 : SUCCESS in 53m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/542a5e8fec8a4cad8c60b22f85a2e848 : SUCCESS in 27m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e7f13e0430104ff1a95cb322e92ffe47 : SUCCESS in 17m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a70bba98896749fe8982b227eb751d55 : SUCCESS in 55m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ac1f41757b44a8a5216a6d329052a3 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba67a68179243da8eeb9baecbf06f13 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee1fd7457e24693ab51fdacc3e01e97 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8751daa7f71415f8453c2e1e347d8a4 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1dedaf37c34d06b264a5a1b14669da : SUCCESS in 4m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a23c0bbbadd64f4ab301f6f8460085d7 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99b26cfa1afe4d2aa0377df5a00cfe42 : SUCCESS in 25m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86a5fa306737425cb65748b31819852c : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/86a7ebd6914f412b9c20ef4eeb8d4b06 : SUCCESS in 54m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab377abcf60f4399859522da6e816247 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":1},{"id":"67e853f3f6b61a4ae2f4a8d59d7b6059da2c5247","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 13:06:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1524905ee7976472168f5a6d06edc5911931b892","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-11 13:08:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"37c65b270206fc1eed4076e5b6bd6a0cf5b74d9b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-11 13:41:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/aec87fc350c74730bcd991a4ec8f3df4 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5ba2c3781fa9935fe310ca7ea356b5a45f4cdfaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 14:41:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bbf2935bdb234db3bbfb90b5d81f4cfe : SUCCESS in 36m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/972efc05c3664699bbafd9ad11d6f657 : SUCCESS in 41m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/21cd8efecbbf4464b6ed830fa797fec7 : SUCCESS in 1h 16m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d6d4335239ce4dc2bcd4f7ec161a9bf5 : SUCCESS in 1h 08m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/446a645b370746bcbd2b4c65232f17a1 : SUCCESS in 1h 33m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cb7d16531945440cb4fafb11f26e17a8 : SUCCESS in 53m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5efe80b5fabe47809226fa1f179295e5 : SUCCESS in 31m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f305705a74614f1596d8fba54331156e : SUCCESS in 30m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b2d6734c08041b786cfa5f7a4549008 : SUCCESS in 53m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51034ee3d203434fa380b019687d4802 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ca7b63246444c48249a965b83f4ea5 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5894e63fed04252b802b22ec17ce18a : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd0fe0a2ef7047ca9b6117e548ecd85a : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb309e94904f4c94b57ade51417dc429 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac7fb8be1542451da8b2518ccde7941a : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b346c9eac7d41b0b23388839a750428 : SUCCESS in 34m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3107ba5840bc403d8f67b566294a39e1 : SUCCESS in 25m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9aa5c746b46342419365b4cfa99174c1 : SUCCESS in 58m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c31f55c2614743b5a4269f1b54b6d867 : SUCCESS in 47m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4ac2632f1c44db8ac6d5c25dab9b523 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":2},{"id":"778551fda34d883a1d85a7ac58004f67571ff54e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-11 14:56:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"236bff977fd9e7bbab00f1c9f488b8e1c447ac66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 14:56:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4465c8bcb5ae73a088607e387de65b63f2f7459b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:41:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/99b36c94477441f4b467167cdc31e8b8 : SUCCESS in 51m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1c13b2e4eb074d50b164b54d1ebf4acb : SUCCESS in 1h 29m 08s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/173be3d50ded4d6d8fd7d5d552111602 : SUCCESS in 1h 41m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/65cff41664744173a2ce3a5e3460deca : SUCCESS in 48m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15c876f66952424e921a10152f3699db : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5cb2fbceb454e0888c0a7632833bac1 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4e0e1f3d52e46a3902703c2946e9750 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/467e347884cd4e8d8614a558a5cde27d : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37134e0da5824e2c945a1662fc81144c : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a0ffcaa8a7445c3b66cb7cc7b20c81d : SUCCESS in 5m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11ba21bee96a42198ce67ed392b780d7 : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b60acff064074b76b7ce4f7324985f13 : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ec6d8545d4c40bb87f3701952b63649 : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2ee3afb19989a9ae2d8038fa7341a97f274555a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:41:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1688d4bdc3816b297dc0eb827accf0af9da1db6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:43:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6cdc8668f84b4ca2f70300883b8cda : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/345058b8828049a38a342fe63a2d9967 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c3d567d33931213ac862e04a740bcf6d7e72c35","revisions":{"f4dd5db41ca4cbc07a10aa2cdb475bcde9d3db24":{"kind":"REWORK","_number":1,"created":"2019-12-11 09:33:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/30/698430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/698430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/698430/1"}}},"commit":{"parents":[{"commit":"888833a7f06fde36ad90d6222e1c732ad48761cb","subject":"Merge \"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/888833a7f06fde36ad90d6222e1c732ad48761cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:30:29.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:33:13.000000000","tz":60},"subject":"skydive: mark as unbuildable on !x86","message":"skydive: mark as unbuildable on !x86\n\nIe586b0f249e545877ab1dd156d08e1863cd479ba removed use of rpm packages so\nit is time to disable building for non-x86 platforms.\n\nUpstream provides own container images for several architectures.\n\nChange-Id: I4778db57dae0953f5c7116105466c2830592b514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4dd5db41ca4cbc07a10aa2cdb475bcde9d3db24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4dd5db41ca4cbc07a10aa2cdb475bcde9d3db24"}]},"branch":"refs/heads/master"},"3c3d567d33931213ac862e04a740bcf6d7e72c35":{"kind":"REWORK","_number":2,"created":"2019-12-11 13:06:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/30/698430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/698430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/698430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/698430/2"}}},"commit":{"parents":[{"commit":"888833a7f06fde36ad90d6222e1c732ad48761cb","subject":"Merge \"Revert \"Introduce macro to remove Py3 only plugins on CentOS 7\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/888833a7f06fde36ad90d6222e1c732ad48761cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 09:30:29.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-11 13:06:33.000000000","tz":60},"subject":"skydive: mark as unbuildable on !x86","message":"skydive: mark as unbuildable on !x86\n\nIe586b0f249e545877ab1dd156d08e1863cd479ba removed use of rpm packages so\nit is time to disable building for non-x86 platforms.\n\nUpstream provides own container images for several architectures.\n\nChange-Id: I4778db57dae0953f5c7116105466c2830592b514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
