)]}'
{"id":"openstack%2Fkolla~698855","triplet_id":"openstack%2Fkolla~master~Ied4a00e8866560d39004a34d6857ac62e22e2c0c","project":"openstack/kolla","branch":"master","topic":"wip/infra/unify-centos-base","hashtags":[],"change_id":"Ied4a00e8866560d39004a34d6857ac62e22e2c0c","subject":"base: unify CentOS packages list","status":"MERGED","created":"2019-12-13 09:29:50.000000000","updated":"2019-12-16 12:24:57.000000000","submitted":"2019-12-16 12:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698855-1576498894094-0548d2f5","meta_rev_id":"95da8942b0e3a4a57603000bceba39da2d957fb3","_number":698855,"virtual_id_number":698855,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 12:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-13 11:18:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-16 10:35:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-16 08:16:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-12-16 10:35:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 09:30:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-13 09:30:40.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2019-12-13 11:18:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-16 08:16:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-16 10:35:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-16 12:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45e21afb7fc899c8ca4e18532952b03fb6c46e47","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 09:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c559aa67bcc0dfa63fdd731dc25a987d42048271","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-13 09:30:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"629e9faaf7f308393e3fab7c8726a032e79bdefa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 09:47:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9da4149166c00807f18970fb9a1b0b4e95b62fdf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 09:52:11.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nAlso, release note, including any deprecations and/or breaking changes.","accounts_in_message":[],"_revision_number":2},{"id":"a0eda4cfa1d20610148f023146cbacf19a7cbb7c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 09:57:33.000000000","message":"Patch Set 2: Workflow-1\n\nas Mark wrote - some work needed","accounts_in_message":[],"_revision_number":2},{"id":"d89c3cf2508b5ed6d2f1d1d6c3b18c85fcc899e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 09:59:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c0da9bbb0e467fa35cf1e290fd8f0bd846be71a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-13 10:02:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/597e712c20764669acab698bda05aac3 : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0b990d294f607c53cb9ea2431591010767136c1b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 10:08:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1b85d4cedaae9eb1f193a7e4d63b55194cd4c812","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-13 10:40:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/9964bff1ecab49c9882a335b065e95ca : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":3},{"id":"36c709905f0994f7e5f83dca2ea8a70273c103f5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-13 10:41:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a11feb8d767dd5c31a6734215ec90a652f81aac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-13 11:18:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7d67297fe16f46479955036cc0b73823 : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c552acf6c784ece067d6aae1847b4189a5fb7f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:13:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5462406937d846689f7484f55e0145b4 : SUCCESS in 31m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a915b318b57c4e6f85946a897656fea9 : SUCCESS in 38m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e64d1049f8ec4559a03abb77ad84055f : SUCCESS in 1h 36m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b2dcaa44a7184a70a19bfbaf9db93584 : SUCCESS in 1h 37m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b01c89141ff3493ab285a84c24c29793 : SUCCESS in 1h 46m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ab9a1c6376e54a96bceeff0149131351 : SUCCESS in 53m 29s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/3b86cff91085434383afe04934d2c09c : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/1a70fa8e0c0649309f7ce0ac9f3ca145 : SUCCESS in 2h 11m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2059652d4a984fdcb8b67cb33574fed5 : SUCCESS in 2h 26m 44s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/a7b29a3302fa4ba4adeb01436510faa8 : SUCCESS in 1h 48m 04s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3a537d56f41448e2ab816d3367126ef7 : TIMED_OUT in 2h 02m 50s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/af150d96a2a34a879c6fefd6cd32dd70 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dab958acbe4d4b0797cba2686e0f3d99 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e92cf46459d4cacb1c40590aa9e5bc8 : SUCCESS in 17m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39e91dc5acb9482688f4747b81b53688 : SUCCESS in 57m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2579d0dc3fc04c7b88e88ac448895aa2 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2344effee648f19d003c5624d642a5 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03141a43598a48a1a15b2c12fa5c60ae : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfdba8dece9c47df9a55b98503a70b26 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fb396387a349ba8167bd1836298c54 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fb10e1fad8d48949e4e222da557505f : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57927d34251c41f29c316aee8a98426c : SUCCESS in 25m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f41b497393c444f9244a059eaf5a88b : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f8805abeffb4011ac60ac2f913f6b24 : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd75ebbb43e848ac883074ff57fc27c2 : SUCCESS in 54m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee2ed9260aee4c3ab903d35811886721 : SUCCESS in 1h 03m 57s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/eaf6c967e0de435b9f617159ba80d760 : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0c476c8bd6244ffebc27ee56451b3bde : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/23fe32493e1a4dabbca4831af1db7bc8 : SUCCESS in 1h 50m 00s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/bab8a1ec40214e0f9d57b13c508c5b5a : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5f54cd919f7c4a30af41a5cf37e3e9d9 : SUCCESS in 1h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8cc5b9393292b608d45ddc9c60daebdcf6dac6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-16 08:16:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ceec115a513027f2b732b61d69e2f5c9ac371dca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 10:35:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\ntyvm","accounts_in_message":[],"_revision_number":4},{"id":"ac201ea4e7799d07f4c5c110d4e4df950be9859f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:35:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fea468b20329cca8edde5ed0ed7575bbc9ea5029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 12:21:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/39755919c47345b7bcebc03b567b9ce4 : SUCCESS in 46m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f130c9bbca11473298b4f2599668f1df : SUCCESS in 1h 16m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/276b313ad1d14855b7f1db958e88dafa : SUCCESS in 1h 39m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6cbce0e01a7248738e287e38dd923ba1 : SUCCESS in 51m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d6f40e69f37485d804eef574af59ab2 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98199a834f32477a993200ea174bd32f : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e64cb3575a4342abd310e67c66c913 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59370e37088a45a8856fcb0492b79abf : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db21bc6b017f4f45ab46a300435e0ead : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a6e0ca730d14c1ea3d0067087366090 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4968a88866d844e790893973debdb458 : SUCCESS in 30m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb11f57caea54f83852af61281e1df1f : SUCCESS in 58m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc8299593824441cb54d902ce00d5d38 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":4},{"id":"89b491132fb748362b5f5854a6686ab53844ebb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 12:21:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"61ac02b91eff3adb4985b67168cad4f8a1a5f24a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 12:24:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b2f6acb40f445cbabe519af8d4d56a : SUCCESS in 2m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc0492b1e8b447a3a78d5003e1698f53 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e941562d948c98735e4b2074025c89dba436c89e","revisions":{"e9cf8aa4777c4a5cd5f741fa51abc814fd6360e2":{"kind":"REWORK","_number":1,"created":"2019-12-13 09:29:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/55/698855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/698855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/698855/1"}}},"commit":{"parents":[{"commit":"3c3d567d33931213ac862e04a740bcf6d7e72c35","subject":"skydive: mark as unbuildable on !x86","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:20:25.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:20:51.000000000","tz":60},"subject":"base: unify CentOS packages list","message":"base: unify CentOS packages list\n\nChange-Id: Ied4a00e8866560d39004a34d6857ac62e22e2c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9cf8aa4777c4a5cd5f741fa51abc814fd6360e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9cf8aa4777c4a5cd5f741fa51abc814fd6360e2"}]},"branch":"refs/heads/master"},"9c0df31aa9df0ec2a547272bb4aab99358351e23":{"kind":"REWORK","_number":2,"created":"2019-12-13 09:47:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/55/698855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/698855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/698855/2"}}},"commit":{"parents":[{"commit":"3c3d567d33931213ac862e04a740bcf6d7e72c35","subject":"skydive: mark as unbuildable on !x86","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:20:25.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:47:20.000000000","tz":0},"subject":"base: unify CentOS packages list","message":"base: unify CentOS packages list\n\nChange-Id: Ied4a00e8866560d39004a34d6857ac62e22e2c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0df31aa9df0ec2a547272bb4aab99358351e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c0df31aa9df0ec2a547272bb4aab99358351e23"}]},"branch":"refs/heads/master"},"8f619dfae376a7b65e45bec566096b5c3bdfa9c7":{"kind":"REWORK","_number":3,"created":"2019-12-13 10:08:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/55/698855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/698855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/698855/3"}}},"commit":{"parents":[{"commit":"3c3d567d33931213ac862e04a740bcf6d7e72c35","subject":"skydive: mark as unbuildable on !x86","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:20:25.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 10:07:31.000000000","tz":60},"subject":"base: unify CentOS packages list","message":"base: unify CentOS packages list\n\nFor infra images work we need \u0027base\u0027 image to be \u0027install_type\u0027\nagnostic. So it will be the same when built for \u0027binary\u0027 or \u0027source\u0027\ninstall type.\n\nChange-Id: Ied4a00e8866560d39004a34d6857ac62e22e2c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f619dfae376a7b65e45bec566096b5c3bdfa9c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f619dfae376a7b65e45bec566096b5c3bdfa9c7"}]},"branch":"refs/heads/master"},"e941562d948c98735e4b2074025c89dba436c89e":{"kind":"REWORK","_number":4,"created":"2019-12-13 10:41:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/55/698855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/698855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/698855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/698855/4"}}},"commit":{"parents":[{"commit":"3c3d567d33931213ac862e04a740bcf6d7e72c35","subject":"skydive: mark as unbuildable on !x86","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3d567d33931213ac862e04a740bcf6d7e72c35"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 09:20:25.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-13 10:41:40.000000000","tz":60},"subject":"base: unify CentOS packages list","message":"base: unify CentOS packages list\n\nFor infra images work we need \u0027base\u0027 image to be \u0027install_type\u0027\nagnostic. So it will be the same when built for \u0027binary\u0027 or \u0027source\u0027\ninstall type.\n\nChange-Id: Ied4a00e8866560d39004a34d6857ac62e22e2c0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e941562d948c98735e4b2074025c89dba436c89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e941562d948c98735e4b2074025c89dba436c89e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
