)]}'
{"id":"openstack%2Fkolla~699361","triplet_id":"openstack%2Fkolla~master~I7a92b762f6a783f92ebca61e5595e1422a92df4f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I7a92b762f6a783f92ebca61e5595e1422a92df4f","subject":"build: list only buildable templates for --template-only","status":"MERGED","created":"2019-12-17 07:50:25.000000000","updated":"2020-01-28 10:46:37.000000000","submitted":"2020-01-28 09:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699361-1580204977028-ff1353d7","meta_rev_id":"acbf2de15ebbc07e0552cad8cbaeba65e3ec029a","_number":699361,"virtual_id_number":699361,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 09:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-27 12:17:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-28 03:09:32.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-28 08:00:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-01-28 08:00:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 20:20:16.000000000","updated_by":{"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},"reviewer":{"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},"state":"REVIEWER"},{"updated":"2020-01-08 16:00:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-08 16:00:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-27 12:17:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-28 03:09:32.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-28 08:00:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-28 09:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2308c413b0cf2a4203068c2b94c8deba2a8659b3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 07:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7938e93ef60548880ee75e7e105f3a7bb541f5e6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 07:52:31.000000000","message":"Patch Set 1: Workflow-1\n\nneeds switch to enable removals","accounts_in_message":[],"_revision_number":1},{"id":"770359755b308225d97a2a616a0898e6314d3289","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 08:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/0ee81195d0494824b6352c1e294071e4 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"af871803031551839db2c222eacfdf32240bf2c4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 08:25:08.000000000","message":"Patch Set 1:\n\nWould it be hard to limit file creation based on our filtering?","accounts_in_message":[],"_revision_number":1},{"id":"9ccbd87270aa62108171f560fca64e289b6f0518","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 08:28:16.000000000","message":"Patch Set 1:\n\n\u003e Would it be hard to limit file creation based on our filtering?\n\nWe create all docker files then use them to build a list of images then traverse through list converting it into tree, then filter tree to contain only things to build.\n\nThen we remove those docker files which we know that are not buildable.","accounts_in_message":[],"_revision_number":1},{"id":"830c825bd199558d8bd7b48fc9c41b2a10dd560f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 08:53:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ea521c93e7c40109f63795dd69f33fd : SUCCESS in 1h 01m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"994637dbd8754398d40cd1ff5c24cf0113ea9f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 09:41:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/82c3d50851664181aec53252474353af : SUCCESS in 49m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/39ab41717db648e99bc2c4058ee9932c : SUCCESS in 43m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4713fa6c503e4e719ca14490169cefcf : SUCCESS in 1h 07m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/aa0f6bdc95304471ac1d5b6e8b47526c : SUCCESS in 1h 47m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7ab2e4aa742f479e9dd4a5479ad07a6d : SUCCESS in 1h 45m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/173e53120376436d940e2db633130b39 : SUCCESS in 1h 11m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8af4bfa42f51482fb8053f62a5d40b29 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/44b3fc0e165b47c2bf25e5149d3f2a05 : SUCCESS in 43m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33f64da53f014c5792de9ec5b0af7ec7 : SUCCESS in 1h 09m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a086c8b9854021854cc1cd767c7078 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07ab8cf49d34d0b9467703353e8bbb3 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/263fe3e0508744649a84c3f7bcb66590 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69598f169f5d4a1bb782f716c777ee4b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b40e4359ef4253a953226ea5ef12c0 : SUCCESS in 7m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca96163483cd4a8786608220c5ac3e78 : FAILURE in 37m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d3bc21f57584c1881ad209436d1890b : FAILURE in 32m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87c9543af6554fae9d5ff34d08d0fb23 : SUCCESS in 1h 12m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc54da80c839421d8c333cb65ddb8a89 : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4eb11fb989244ee85a6c0b3127d9a87 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"13384d8dab809523e8f4070b4bf0dbf30565bd57","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 10:28:44.000000000","message":"Patch Set 1:\n\nMakes sense.","accounts_in_message":[],"_revision_number":1},{"id":"36a2dfdfad882f955582fdf7a8dfcb53918f6193","author":{"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},"date":"2019-12-17 14:52:23.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m not sure I understand the intent of this patch. Doesn\u0027t kolla already skip unbuildable images? [0]\n\nWhat we\u0027re trying to overcome in https://review.opendev.org/699024 is an inconsistency in what containers are attempted to be built. For example, one run we\u0027d skip skydive and succeed [1] and another build we\u0027d try to build skydive and fail [2]. I\u0027m not sure why this is different from run to run.\n\n[0] https://github.com/openstack/kolla/blob/master/kolla/image/build.py#L1381-L1387\n[1] https://centos.logs.rdoproject.org/tripleo-upstream-containers-build-master-ppc64le/1837/logs/logs/build.log\n[2] https://centos.logs.rdoproject.org/tripleo-upstream-containers-build-master-ppc64le/1836/logs/logs/build.log","accounts_in_message":[],"_revision_number":1},{"id":"694018a0db4e95510129e573b7b12e09b612bc6b","author":{"_account_id":11929,"name":"Michael Turek","email":"mjturek@linux.vnet.ibm.com","username":"mjturek"},"date":"2019-12-18 20:20:16.000000000","message":"Patch Set 1: Code-Review+1\n\nHad a discussion with Marcin, I see the benefit now. Basically in the case of buildah, where you\u0027re only generating the templates, this would prevent unbuildable images from getting built.","accounts_in_message":[],"_revision_number":1},{"id":"26e52d6816fd913de0dab186d963fa63988adb46","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 12:39:23.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"41daaca9728e884e50797cd5f606071f6c68e872","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 12:45:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3411481d6802f4b0c82c2e74a8c09eee119fc7cb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-07 13:11:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6e5bc47401854d14b66aced7eb1ff899 : SUCCESS in 24m 27s","accounts_in_message":[],"_revision_number":2},{"id":"45fc84b5507f4abbdbb535fb467989c42b52f4a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 14:58:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c691d4c0f0043e49f5c41641c183813 : SUCCESS in 2h 09m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00a2a8b6111c64b10e61d0ff10488bcfc93e764f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:18:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d95f18db90c24231b830bdcaa2d51383 : SUCCESS in 49m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c8f05bc9608342c0bee29f51b22befdc : SUCCESS in 38m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4e24c2ddac554183bd4ddc93f2a75374 : SUCCESS in 1h 35m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/530ef91f55ab4aed99cf3be5af1090dd : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c2cb31b6a7de40e483865d8e8e6bd48b : TIMED_OUT in 2h 32m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/730c257300534d96bbb7e8735c87a65b : SUCCESS in 57m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9ea0235646db48ddae410b6dbc3ed577 : SUCCESS in 43m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a6bb1ecbcb44d07beb2ffa7e83eadc6 : SUCCESS in 21m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39b4a081b5ce4d5eb7d183a13fec3b4a : SUCCESS in 57m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/948e53e44244467aa66d5148c6279648 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf32d8a37f3f487a89956470fe949b75 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b534ab53cce243fb977b45cb4165c0a1 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b47980156584892a5de3fe54826300c : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ee96691fa94a90a82251697213a95d : SUCCESS in 6m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2150577da8584dc0ad23e498905b10ff : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/acb75bf02ad6461a94a1d781fec841cc : SUCCESS in 38m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46f87ad3f2f14b12900235c70568adc0 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e33932c93e6e483e857194441a7ba5ff : SUCCESS in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8aba0438c99446ca3ce32eae26d029e : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":2},{"id":"89d3cc91558d97c3a9ff48a01c8312c58b3b7670","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-08 16:00:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"637d0fbe26b2d65621c43eaa9019293fc3604362","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-08 16:27:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/dd4b0cbe8b4f47109ca50c0f7b287b84 : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d7a4681783c46eb6ce4db7450e1c6dfeecf093eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 17:46:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/33557eee004d4a5884c0409c2c54955b : SUCCESS in 37m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/72fb7e5a356e42da9048db17857b08f8 : SUCCESS in 46m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/481522b9fd5a4aaaa28e1befdba5d998 : SUCCESS in 1h 33m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d3722cdc242a47fcbc3b3cf374ea6139 : SUCCESS in 1h 18m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ff170d192a54eb1ab1c260d7f0aacec : SUCCESS in 1h 40m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8fb0a98f7ac846839bdc39d80f445387 : SUCCESS in 1h 00m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e391de8948de4021a3639f2f2dd6b115 : SUCCESS in 27m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/510eeb3cd0a347e78361254755397115 : SUCCESS in 19m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9d824138cee141efb80012cfd5de5019 : SUCCESS in 1h 23m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4eed1d55cf043fa9299dbc0fdfcf365 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8377d8ba10974390a4d6ba31ab0c0265 : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d1968b7593841bc9e14f34f6b7e6fc8 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c7fbe2b88e543e7a341003bd4810821 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bace740587d4cf8b562e5fed67958e0 : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/002e186c0fb44ea18defcae44150e037 : SUCCESS in 36m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/449828c6f0864b8483e609db768e8286 : SUCCESS in 35m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/884fe6bb73684154b1f0883282b761a2 : SUCCESS in 1h 00m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5054288230244d3c949c488a77105c53 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d335dc756824dbfab20679a1d2a0b1e : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":3},{"id":"72a9b29558c6fb3008f8506ce35d053ca0752931","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-08 18:35:22.000000000","message":"Patch Set 3: Code-Review-1\n\nthat NOTE from the commit message looks important","accounts_in_message":[],"_revision_number":3},{"id":"a474a22e5ae09ea6472ae953c997d01488e57e9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 18:38:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f214a7cddd264e0d96c27a2b131d8934 : SUCCESS in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3559e761f9768634ef2f9ed6ccb4354a0391467e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-27 11:47:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e1f952fc6466e6a6ae9168dbe4ccb89051438ca7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-27 12:17:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f24f6f5358f143a9bfe4b134e6b33435 : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ade87d77fda2f256347a713a2f623c3c36c71682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:27:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3243a7839c19425da958d81f5086eb62 : SUCCESS in 33m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8fd7dd90d9854929b8adf63c71fbb0fa : SUCCESS in 1h 01m 38s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c32f5a2f2422435598ca3ce546575bde : SUCCESS in 6m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/553a0c7b3ac24ed4b7f66d775a19adba : SUCCESS in 5m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e57df1e458e84ef1bb7df8605e9ed037 : SUCCESS in 1h 25m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/45cca239ed034c0d9179e3ed2343a1c7 : SUCCESS in 1h 20m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2dae145724b14d05a32f4f85e47bdf1b : SUCCESS in 1h 26m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/54d771963d3c420fbf2922800548ccf3 : SUCCESS in 1h 00m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6e1825557f44b3fbbc47ec2ec31928b : SUCCESS in 45m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f483aec11b1436db2635e8479d36158 : SUCCESS in 21m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e32540dd9e724464a2967b940c517210 : SUCCESS in 1h 31m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90308c50b0e244c594fbea9e8b70d9db : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b624efeb234a549255f4addc538e81 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9e0a3488bcc45d3a539c0bc4a51fc99 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bf74f1280f14007bb7f4a8eec5c8ad8 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a157822eacb4430b623bc849f6db97e : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f102e3ceabb6449d8525d4413fa14d77 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/560f3832edfb4cc4b5e22a0442bb13e5 : SUCCESS in 35m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f47572fc12748a79e55e3dbe11b601c : SUCCESS in 24m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4015e68e980469ea1730413fb57d1c1 : SUCCESS in 56m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47ef000b9a1e45d3bfe6b9725276c57d : SUCCESS in 1h 32m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b72999c41544ffc9354735997bf034d : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":4},{"id":"b61d3858a88e530ae5b97fc28d274c09cc6adc22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:34:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f532fbbb458242bf866af84be05fd27d : SUCCESS in 1h 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b6e1be80c9ac63df406f97b8aa865b829defd5a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-28 03:09:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3fdd8518ef1fec73cfd0ccdc2f9c2e94d9779e88","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-28 03:09:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b7dbc3ff0f20ea808898700f56b557f6a325c049","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 08:00:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b42b72d1cefd004a68c4a0c89e688c19f746b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:00:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"763a867d4f6fbb060a6fcaf32efc50107fbf9476","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 08:03:20.000000000","message":"Patch Set 4:\n\nneed rn in followup","accounts_in_message":[],"_revision_number":4},{"id":"e1401bda97fbb448c1502e269573617e8ab2e0a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:49:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f9ed9e513d0a4d2eb11587ea6d55527f : SUCCESS in 39m 21s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a13a4a43f11c49bd8d8224d88f07bac0 : SUCCESS in 5m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ad48f26fdbe24336aa3d26a49068cbcd : SUCCESS in 1h 29m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36b15ce9543a4bcdb33e975a012b80b2 : SUCCESS in 1h 25m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9fc29726c16240f684cd069b2833b7c9 : SUCCESS in 43m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af9f3347a111445e8c01f0c71c20f7c6 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e304d781614670b9d9b79b133b9502 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d260f12b4ab54610b8484d8db3bbf9a5 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f4a3b7b2c6f4e3caa789ee3c556e299 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae647b8fa914431ac14f389c7276ae1 : SUCCESS in 6m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e3d7411fd14464ab661fbc085ba362e : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55521c95b4e24c6397e4353596899184 : SUCCESS in 1h 48m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/418a853329c84f1a9b8cb573b89bb576 : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":4},{"id":"93f8e4c272d234343062d1a6215c0a708b70b129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:49:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5a7ada47c44e2d1366a63f654181c6c745d3e35d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:51:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe0a4a43e594f489805c189c241e3ba : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4},{"id":"5db803ee6e86a5b0105dd24e408fd4a0e4b98693","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 10:23:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a95b00012c01cfd15a040530eb993ef76f28fcb5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-28 10:25:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"acbf2de15ebbc07e0552cad8cbaeba65e3ec029a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 10:46:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","revisions":{"b2c2c00c65fd3478070caf471727d98a031ad75c":{"kind":"REWORK","_number":1,"created":"2019-12-17 07:50:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/699361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/699361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/699361/1"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 07:47:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 07:47:42.000000000","tz":60},"subject":"build: remove not buildable templates for --template-only","message":"build: remove not buildable templates for --template-only\n\nTripleO uses --template-only to generate Dockerfiles and then use\nbuildah to create container images.\n\nBut as Kolla generates all files then they need to have a way to exclude\nnot buildable ones. https://review.opendev.org/#/c/699024 is one of such\nattempts.\n\nBut if we remove not buildable Dockerfiles then they get what they can\nbuild.\n\nChange-Id: I7a92b762f6a783f92ebca61e5595e1422a92df4f\nNOTE: some switch to enable filtering probably would be nice to add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2c2c00c65fd3478070caf471727d98a031ad75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2c2c00c65fd3478070caf471727d98a031ad75c"}]},"branch":"refs/heads/master"},"76e48ec5ae3862823343fee7dd2534791b724095":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-07 12:45:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/699361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/699361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/699361/2"}}},"commit":{"parents":[{"commit":"8036dd4848b1e428ea940183219a3f63c4635560","subject":"horizon: fix build in CentOS7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8036dd4848b1e428ea940183219a3f63c4635560"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 07:47:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 12:45:15.000000000","tz":0},"subject":"build: remove not buildable templates for --template-only","message":"build: remove not buildable templates for --template-only\n\nTripleO uses --template-only to generate Dockerfiles and then use\nbuildah to create container images.\n\nBut as Kolla generates all files then they need to have a way to exclude\nnot buildable ones. https://review.opendev.org/#/c/699024 is one of such\nattempts.\n\nBut if we remove not buildable Dockerfiles then they get what they can\nbuild.\n\nChange-Id: I7a92b762f6a783f92ebca61e5595e1422a92df4f\nNOTE: some switch to enable filtering probably would be nice to add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76e48ec5ae3862823343fee7dd2534791b724095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76e48ec5ae3862823343fee7dd2534791b724095"}]},"branch":"refs/heads/master"},"22cdb0d3e4171a1b23c7bff6ebfbf1a494274c5a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-08 16:00:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/699361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/699361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/699361/3"}}},"commit":{"parents":[{"commit":"8b7ec43536842dd500cc5ee1198a318070ac1046","subject":"horizon: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b7ec43536842dd500cc5ee1198a318070ac1046"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 07:47:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-08 16:00:31.000000000","tz":0},"subject":"build: remove not buildable templates for --template-only","message":"build: remove not buildable templates for --template-only\n\nTripleO uses --template-only to generate Dockerfiles and then use\nbuildah to create container images.\n\nBut as Kolla generates all files then they need to have a way to exclude\nnot buildable ones. https://review.opendev.org/#/c/699024 is one of such\nattempts.\n\nBut if we remove not buildable Dockerfiles then they get what they can\nbuild.\n\nChange-Id: I7a92b762f6a783f92ebca61e5595e1422a92df4f\nNOTE: some switch to enable filtering probably would be nice to add\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22cdb0d3e4171a1b23c7bff6ebfbf1a494274c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22cdb0d3e4171a1b23c7bff6ebfbf1a494274c5a"}]},"branch":"refs/heads/master"},"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-01-27 11:47:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/699361/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/699361/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/699361/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/699361/4"}}},"commit":{"parents":[{"commit":"7a57a4230a496250725ae6cd6012eea2a12fe799","subject":"Merge \"Update cAdvisor to latest release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a57a4230a496250725ae6cd6012eea2a12fe799"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 07:47:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-27 11:46:37.000000000","tz":60},"subject":"build: list only buildable templates for --template-only","message":"build: list only buildable templates for --template-only\n\nTripleO uses --template-only to generate Dockerfiles and then use\nbuildah to create container images.\n\nBut as Kolla generates all files then they need to have a way to exclude\nnot buildable ones. https://review.opendev.org/#/c/699024 is one of such\nattempts.\n\nBut if we remove not buildable Dockerfiles then they get what they can\nbuild.\n\nChange-Id: I7a92b762f6a783f92ebca61e5595e1422a92df4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
