)]}'
{"id":"openstack%2Fkolla~699395","triplet_id":"openstack%2Fkolla~master~I4d2a6136e4c4501e9f16a145b616fbea0530a966","project":"openstack/kolla","branch":"master","topic":"sensu-client","hashtags":[],"change_id":"I4d2a6136e4c4501e9f16a145b616fbea0530a966","subject":"Fix building sensu-client by pinning network-checks version","status":"ABANDONED","created":"2019-12-17 11:38:31.000000000","updated":"2020-07-13 09:52:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8c63f7b741e23b8f3dcde3e1fa3f0a712772886","_number":699395,"virtual_id_number":699395,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-09 10:10:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-12-18 13:47:22.000000000","_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"date":"2019-12-17 14:24:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-17 13:15:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-17 12:04:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 12:04:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 13:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 14:24:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-09 10:10:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"3317e11b081473bade2077e16796bbd1572f2235","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-12-17 11:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a6280e528508983f2cd08d28a8fd55ce92bbbc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 11:52:30.000000000","message":"Patch Set 1:\n\nStrange that we don\u0027t see this issue in upstream CI.","accounts_in_message":[],"_revision_number":1},{"id":"7b7ded75c90d18787738579f1e8a3197a094e2f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 12:04:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/2a134bb047fb4359aa226fdd65e5be07 : SUCCESS in 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2dacea118f8eb032162ef4eca8e0da4026cf3419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 12:53:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f4c099a479c4afba803fa862ceab1aa : SUCCESS in 1h 14m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89d5b479cc81aca25b9027e9d6408da152e314fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 13:15:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/999b51193a4f4acb8ec1cf154cdf007c : SUCCESS in 34m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3f762bedac2242f6b6444c0293e85c93 : SUCCESS in 47m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f2a5393f09fd47039be8f230c6aadd2d : SUCCESS in 1h 19m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f9ccf25c04e1424cb6b35815426b6362 : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0dec2694305b4008b95ac388c46d0a70 : SUCCESS in 1h 33m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4945adaec4f2437986b9001be407bc28 : SUCCESS in 48m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3bac97f163374547a7b22f57e3778fa2 : SUCCESS in 31m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6a01ca4ec740408fae04f8b2bd09a371 : SUCCESS in 25m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd3957d3042442ab97e8763b7dd08a61 : SUCCESS in 56m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c48d8cc6924699b2c54b754490cfe0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/873229b8bf174289a5b8b882ea52a6dc : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd7e042cf44243cca564da6880370486 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ee703eb64834f528488ca2a39a68a14 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46570ad6c1a7446b922109b372a002ef : SUCCESS in 7m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8b457d1725c4337a7fa791e6f01ed1c : FAILURE in 25m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4299e3290a354ac395d7245e278ffbd7 : FAILURE in 25m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cd1ac9020f04a7eb8ca7f384c9fb945 : SUCCESS in 57m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ed7b660b670400b8e5c6ab27caa2f3c : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c22849cc4034d25a4d30cea5bb1884d : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4762aa5137c1f10a74b0a96673557ed723b73050","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-12-17 13:15:51.000000000","message":"Patch Set 1:\n\n@Mark: Should I share the version of any installed package?","accounts_in_message":[],"_revision_number":1},{"id":"97ce86f98ea0d577a3d8a3a9b11226012e3069d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 14:24:26.000000000","message":"Patch Set 1:\n\n\u003e @Mark: Should I share the version of any installed package?\n\nYes, if you could list the installed sensu plugins.\n\nAlso, which version of kolla did you see this on, and have you reproduced it on master?","accounts_in_message":[],"_revision_number":1},{"id":"70e54518e6f86e8797bf0b65ac155b8b4b4da0e5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-12-18 13:47:22.000000000","message":"Patch Set 1:\n\nIt took me a while to grep more details.\n\nINFO from local host:\n\n    $ git remote show origin\n    * remote origin\n      Fetch URL: https://github.com/openstack/kolla.git\n      Push  URL: https://github.com/openstack/kolla.git\n      HEAD branch: master\n      Remote branches:\n        master        tracked\n        stable/ocata  tracked\n        stable/pike   tracked\n        stable/queens tracked\n        stable/rocky  tracked\n        stable/stein  tracked\n        stable/train  tracked\n      Local branch configured for \u0027git pull\u0027:\n        master merges with remote master\n      Local ref configured for \u0027git push\u0027:\n        master pushes to master (up to date)\n        \n    $ git log -1 --oneline\n    6f4e98c Merge \"docker: fix \u0027Empty continuation line found\u0027 warnings\"\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nINFO from the container run from the failed image:\n\n1) listing of the plugins gave no result (note: the path were detected from running sensu-install in a verbose mode)\n\n        # GEM_PATH\u003d/opt/sensu/embedded/lib/ruby/gems/2.3.0:$GEM_PATH gem list | grep sensu-plugins\n        # echo $?\n        1\n        # GEM_PATH\u003d/opt/sensu/embedded/lib/ruby/gems/2.3.0:$GEM_PATH gem list | grep whois\n        # echo $?\n        1\n\n2) installing ceph plugin in verbose mode - http://paste.openstack.org/show/787720/\n3) listing again\n\n        # GEM_PATH\u003d/opt/sensu/embedded/lib/ruby/gems/2.3.0:$GEM_PATH gem list | grep sensu-plugins\n        sensu-plugins-ceph (1.0.2)\n\n4) attempt to install network-checks http://paste.openstack.org/show/787721/","accounts_in_message":[],"_revision_number":1},{"id":"5b9976af8ee8ad278fe44ad0825b9c6e48f5fdda","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-09 10:10:07.000000000","message":"Patch Set 1:\n\niirc I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2 sorted that out","accounts_in_message":[],"_revision_number":1},{"id":"d926736474fbb46d2f4b537d94de80e2ed00b984","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-13 09:52:38.000000000","message":"Abandoned\n\nSensu dropped","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23186eeabd34b99ca1801191081e974a5779a510","revisions":{"23186eeabd34b99ca1801191081e974a5779a510":{"kind":"REWORK","_number":1,"created":"2019-12-17 11:38:31.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/95/699395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/699395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/699395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/699395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/699395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/699395/1"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-12-17 11:33:20.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-12-17 11:38:28.000000000","tz":120},"subject":"Fix building sensu-client by pinning network-checks version","message":"Fix building sensu-client by pinning network-checks version\n\nWe cannot rely on any versios of network-checks ~\u003e2.\nSensu plugin network-checks of 2.0.0\u003cVERSION,3.0.0 had runtime\ndependency \"whois\u003e\u003d4.0.0\" [1] that results in installing whois \u003e\u003d 5\nwhich requires ruby\u003e2 (doesn\u0027t exist at Centos host) that conflicts with\nfirst condition of [2] and fails the build.\n\n[1] - https://github.com/sensu-plugins/sensu-plugins-network-checks/blob/2.3.1/sensu-plugins-network-checks.gemspec#L36\n[2] - https://github.com/openstack/kolla/blob/9.0.0.0rc2/docker/sensu/sensu-client/Dockerfile.j2#L93\n\nChange-Id: I4d2a6136e4c4501e9f16a145b616fbea0530a966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23186eeabd34b99ca1801191081e974a5779a510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23186eeabd34b99ca1801191081e974a5779a510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
