)]}'
{"id":"openstack%2Fkolla~699399","triplet_id":"openstack%2Fkolla~master~Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff","project":"openstack/kolla","branch":"master","topic":"bug/1856699","hashtags":[],"change_id":"Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff","subject":"Fix monasca-grafana being unbuildable due to old npm","status":"MERGED","created":"2019-12-17 11:59:34.000000000","updated":"2019-12-17 19:19:37.000000000","submitted":"2019-12-17 19:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699399-1576609594145-fead5b33","meta_rev_id":"aad10d859a2f017ea758ebac76fe8c93e5c29518","_number":699399,"virtual_id_number":699399,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 19:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-17 16:30:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-12-17 16:06:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-17 17:38:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-12-17 17:37:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-12-17 17:38:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-17 16:04:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 13:43:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-17 16:06:09.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-12-17 16:30:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 17:37:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-17 17:38:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-17 19:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6c15b414688ebeb0aa9606513ec86c87b9094bf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 11:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c120c22947a8d593c384c6b9600422e58aa4b9cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 12:27:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f6a752ba1a4243e09769bc6139a35558 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"15654face982fdf7f42e928769044c3ddd66e800","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 12:58:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a96430373f53e79bf15bb42f2b5263edde9d3c8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 13:23:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/8263723b3bc242089e4a55a5418b554d : SUCCESS in 24m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6b0e86fcfa08a92739823da8c6fb3e16f3283c9f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 13:42:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"531b66b3aa18cbd46511f8ef1bda1ea4c5c36606","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 13:43:16.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"6de39f68a0d0234f873fe5e0e9a9374cf2ace9d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 14:23:22.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nLet\u0027s see what zuul says","accounts_in_message":[],"_revision_number":3},{"id":"5442ad98c691818e658251fd41f14d108920a52f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 14:29:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a1a984c388144f47a692395082ddef5c : FAILURE in 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e9af29155dcfbf0347f882df9d7048eda7b733e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 14:36:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9494d09e5d98ce234d62f8527dae6223c212ccfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:35:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8bd31a94e6aa4d01a3782dc0fa38e777 : SUCCESS in 51m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1a5e1e0a98834a8bb35b46878085f57b : SUCCESS in 53m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f0be8e05e833428da3bd9e24e84c0c43 : SUCCESS in 1h 39m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2fedf8b72bd45b5882b3f6c96017b42 : SUCCESS in 1h 39m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/61189391284a4f1e87751a3b7c4db60d : SUCCESS in 1h 49m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7f27a298b66d49b5bbd4ddc90f9ef80b : SUCCESS in 56m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/921156b9e3e7459e9e3e8f11b1538bd0 : SUCCESS in 29m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cc86674038fe4e2a9af319ddd6dc8de5 : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77ce626ff6814daf8fb39082712ff2f9 : SUCCESS in 1h 02m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb88aeb5f09f499ba1907da0dd2a980c : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e33ad8de46eb4a6ca9cae7b6ece335d0 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/137adb78aa714670ac9752990d8dc4a4 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/992eddee5c2547ba877ba5d199449f54 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc4c6b8b5c141c1979a0f8cf0e9fee6 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1df2637506441b7b88b6a3acad05940 : SUCCESS in 30m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/803eed97daa24b838fea1f56fc5b3267 : SUCCESS in 34m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5dc11f4dec2b496092edf289730b48a9 : SUCCESS in 1h 03m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b396bca9b35b47b4b9d49baf4065633d : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ce159c81d3e4e85af8b0c4e629fa476 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":3},{"id":"60d093a18d785b211606d3d985438449730439c0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:37:54.000000000","message":"Patch Set 3: Code-Review+1\n\nNice work Radosław. +1 to pin it if we can and thanks for fixing.","accounts_in_message":[],"_revision_number":3},{"id":"d8de17ae5d16248c9ce8ce9356e0b3d7e7baad78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 15:41:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c803088e47113fa0790be68f15982e1947539744","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 15:41:59.000000000","message":"Patch Set 3:\n\njust waiting for aarch64 to build","accounts_in_message":[],"_revision_number":3},{"id":"0789a905819ad10905241f61782312e3f900fdf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:58:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05757ec6b0934973bed8fb75ddacc7aa : SUCCESS in 2h 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6fb709f5a373e8072db13dca8c8ad457cd18480a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 16:01:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29e8213c591e6bbabc5d41642ed6bb3716ebfeca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 16:04:19.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"301076254dfd74e63dfa62a4a91dd5e0364d8b03","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 16:06:09.000000000","message":"Patch Set 5: Code-Review+1\n\nLooks good assuming it builds","accounts_in_message":[],"_revision_number":5},{"id":"fa04646a5d243cd43acfa922129c826fa1c75572","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 16:30:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a34cfc536cde4eb6ae37343609670ef5 : SUCCESS in 24m 58s","accounts_in_message":[],"_revision_number":5},{"id":"9526e5adb05488e7802546b2855c249e87b64184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 17:36:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/def692d555484785a9ae9f51d9bf6a97 : SUCCESS in 50m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e33d3ad7785c411c9d47988f236e09d6 : SUCCESS in 38m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/38a93b7b6962488892f9678d15b3984e : SUCCESS in 1h 19m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/18b027dad75944a1ae6b5b9b5f59282a : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5df3142281f24f58a4229ba526472213 : SUCCESS in 1h 27m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7fb0e6a421bc43f1bed01343e5af6a02 : SUCCESS in 54m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/46305ded9bd541669a51156e375c5eb5 : SUCCESS in 36m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6dc3f88d89e4d94af8c647a4c99b036 : SUCCESS in 20m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/90bda0d02ce8446fb129d7cfc0a8bd59 : SUCCESS in 55m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87dff433956d4fbf9ae815c3d901a7b4 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53647caa601046afabe250100b61565a : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f98e63643e62401ba1198bdd5d47e095 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37feae40b34e40979e3ee5d49370b6ee : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd3568869104e2ab972bb484af3db5c : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a4f0ad3af384e94a1240b3fe8677460 : SUCCESS in 23m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b37ca6eeee94e17823e432edda3a8c3 : SUCCESS in 28m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93879a10fdb54f8a81524c2040657469 : SUCCESS in 59m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6bfc056d6084a0ab5aaeaf4d5aa834b : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31125264679e497baa58d2afa157863a : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":5},{"id":"67cfb9f72ffb06292f3b7923766291b2476bbb56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 17:37:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4048ba555ff3e84027e9e657d6f11880afbb02ad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 17:38:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"14b84a468fe7180ed31a4a75e40e4365ab853c19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 17:38:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f2a99988ca5b5e9cadab0efc971c7f0eb65986f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:03:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0498367c786e41a9a0afeb959657f892 : SUCCESS in 1h 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3917722f8cfc8913baf9b3604222579d67a01513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:06:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1b9108bd4ffa4d5c991056472d2b9dd3 : SUCCESS in 40m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd39fb93de3b4e00bebc238715583b36 : SUCCESS in 1h 21m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/32979c1a5df54c568a03964149d53b60 : SUCCESS in 1h 26m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4e1799ac761045b4bffe18c563faeb19 : SUCCESS in 45m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aff24f430154f01a8ebb53b6d5a3cbe : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd92b02019b3448da4afbfe3b63cd9cd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e113e5a8b14d467da945d7a7a8051b7a : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a6926bbdd7c4117aaa9f6e31c479bfa : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea85e2c4192f46aaa8070abdc7980fdb : SUCCESS in 7m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d26ae2fecb144fe49b0cee1e909ec7a5 : SUCCESS in 37m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bab64909b7fe43c79596bd2eeeacf2c5 : SUCCESS in 1h 04m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95c7ef6fc334417795d643be4c5156de : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":5},{"id":"f585138a97cdcc29efb49c88638b05f41427aea7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:06:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"01676fec14a9b66c9cb4561ce4431d8fcb29de58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:10:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aade50b8fb914eb4864bd615598b2cd5 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"8bec8ccfd524ccb97c54994b24f3eea397e3006f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 19:19:37.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b2e2c68b86b7a7b16ea270f53dde7e400cae1560","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"308e85aa413da84ed107b27fa09a3f0e0a64392f","revisions":{"cd43f789b39b40af01b445917a3aaef7c05ae576":{"kind":"REWORK","_number":1,"created":"2019-12-17 11:59:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/699399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/699399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/699399/1"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"subject":"Fix monasca-grafana building due to old npm","message":"Fix monasca-grafana building due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.1.0\n\nThis patch runs npm update via npm.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd43f789b39b40af01b445917a3aaef7c05ae576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd43f789b39b40af01b445917a3aaef7c05ae576"}]},"branch":"refs/heads/master"},"5a6feb48554bfdd24e2a6f73851cbb98f9df2526":{"kind":"REWORK","_number":2,"created":"2019-12-17 12:58:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/699399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/699399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/699399/2"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 12:56:52.000000000","tz":60},"subject":"Fix monasca-grafana building due to old npm","message":"Fix monasca-grafana building due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.13.4\nAnd CentOS had:\nnpm 5.6.0\n\nThis patch runs npm update via npm.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\nCloses-bug: #1856699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a6feb48554bfdd24e2a6f73851cbb98f9df2526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a6feb48554bfdd24e2a6f73851cbb98f9df2526"}]},"branch":"refs/heads/master"},"215895e510ca9a22ddc2cf253e7b99e0b9bb02a8":{"kind":"REWORK","_number":3,"created":"2019-12-17 13:42:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/699399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/699399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/699399/3"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 13:41:55.000000000","tz":60},"subject":"Fix monasca-grafana being unbuildable due to old npm","message":"Fix monasca-grafana being unbuildable due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.13.4\nAnd CentOS had:\nnpm 5.6.0\n\nThis patch runs npm update via npm.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\nCloses-bug: #1856699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/215895e510ca9a22ddc2cf253e7b99e0b9bb02a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/215895e510ca9a22ddc2cf253e7b99e0b9bb02a8"}]},"branch":"refs/heads/master"},"8fc9daa8f28ece1eb6413fd05f3e94bc97787a26":{"kind":"REWORK","_number":4,"created":"2019-12-17 16:01:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/699399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/699399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/699399/4"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 16:01:35.000000000","tz":60},"subject":"Fix monasca-grafana being unbuildable due to old npm","message":"Fix monasca-grafana being unbuildable due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.13.4\nAnd CentOS had:\nnpm 5.6.0\n\nThis patch runs npm update via npm.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\nCloses-bug: #1856699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fc9daa8f28ece1eb6413fd05f3e94bc97787a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fc9daa8f28ece1eb6413fd05f3e94bc97787a26"}]},"branch":"refs/heads/master"},"308e85aa413da84ed107b27fa09a3f0e0a64392f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-12-17 16:04:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/699399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/699399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/699399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/699399/5"}}},"commit":{"parents":[{"commit":"97446186cf4e08451c44d4d84981b6d9306facd2","subject":"Merge \"Add Debian/source aarch64 CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97446186cf4e08451c44d4d84981b6d9306facd2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 16:04:14.000000000","tz":0},"subject":"Fix monasca-grafana being unbuildable due to old npm","message":"Fix monasca-grafana being unbuildable due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.13.4\nAnd CentOS had:\nnpm 5.6.0\n\nThis patch runs npm update via npm.\nPinned to 6.x.\n\nAdditionally, this moves npm packages prefix (root) to\n/usr/local to avoid conflicts with native packages.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\nCloses-bug: #1856699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/308e85aa413da84ed107b27fa09a3f0e0a64392f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/308e85aa413da84ed107b27fa09a3f0e0a64392f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
