)]}'
{"id":"openstack%2Fkolla~699414","triplet_id":"openstack%2Fkolla~master~Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08","subject":"CentOS 8: Update packages in images","status":"MERGED","created":"2019-12-17 13:44:17.000000000","updated":"2020-01-30 20:15:03.000000000","submitted":"2020-01-30 20:13:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699414-1580415183771-cc57de1a","meta_rev_id":"1f6052ccfb11c23e057239f6940cb3ac360b92f3","_number":699414,"virtual_id_number":699414,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-30 12:46:59.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 20:13:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-30 17:35:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-29 19:09:59.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-30 10:21:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-30 10:21:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-29 14:22:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 19:41:54.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2020-01-29 19:09:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-30 10:21:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-30 12:46:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-30 17:35:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-30 20:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a0831b1c8aad6638519fb59915ece94fa194ff9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 13:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f4ea35f5a3fc9f3f8ec798d912d7952dc87f5ac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-17 14:19:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/cbaf8451dc76440c8f5756d324fcf2e1 : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"141fd6fb0e35744d8a7f7b543d5c4735ee2daea0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 14:33:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"beab39fe010f0f9eb0ba3cb558cbbd69bfe34f4f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 14:57:20.000000000","message":"Patch Set 1: Workflow-1\n\nNo doubt we\u0027ll need some more changes here. This is a first attempt to split some out.","accounts_in_message":[],"_revision_number":1},{"id":"4b9948b40fd827150ab4051d699e17789eb504e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 16:22:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fb6077e2dc28488fb5f4e2f3f17fa04d : SUCCESS in 34m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e6ac48a9985448b0b251ce39e6692d61 : SUCCESS in 40m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d8d145fa6cf94d039bda86de81642e69 : SUCCESS in 1h 40m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a237ce0ac7134e7c8ce2072a089c6d5d : SUCCESS in 1h 27m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0456bbdf3fd94303a88c970bba39d2db : SUCCESS in 1h 32m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/64680ba506794940acc3464ccd6fd73d : SUCCESS in 54m 52s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/890a220136f3483b963a0ebdfcd7c4bb : SUCCESS in 2h 03m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5d088420ecd546198be4c36c5b8e2185 : SUCCESS in 2h 04m 28s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8fd339b07f224ec28ff2ae09a8c6d397 : SUCCESS in 1h 40m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/939d75be275d4b46952ed0b3ab709c3b : SUCCESS in 38m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3a7ed9053f246cf83e7632f652cb38b : SUCCESS in 31m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5da3a9dcdf664d59bc6d9b7394007721 : SUCCESS in 54m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47f3d4a4a8b949db9a1d193d1fd35bb6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95640abc08549a796b232e36d0789dd : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e273e38aea54856b0a03e42951f49ba : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcf877b5f32e416d87c2b29a59d5c066 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31539813a305454e949bd00bd4826eb8 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/63fe0b1a84964ba2886def6dd28dd2f0 : FAILURE in 21m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d93048e839584df2a19bb088dc980949 : FAILURE in 29m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ce6e3d393bb406395bbd1a6db5b860d : SUCCESS in 1h 08m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a74c7f990f324b17b4586c7e02d3c904 : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41e556b722fa48b2bb66af53063faac0 : SUCCESS in 1h 12m 49s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/eb819cf7940e41689ec4c2b7b5b69b73 : SUCCESS in 1h 21m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/60c79034c1e546fbb6fae82103035741 : SUCCESS in 1h 45m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b6724b7afffa4f43bc21d433864974cb : SUCCESS in 1h 35m 48s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fe8dffa65a9b477ea7241d4e475ea975 : SUCCESS in 1h 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"328e9e9f92660db1f7f519ef1411320900d7eb29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 16:37:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2eaaaeed1dca426d826059e715df4d14 : SUCCESS in 2h 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"123d9655f0c1efe981fb538ae5c0a5a0c9ddd9e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-21 14:03:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"de4f314cad798218c178b735f6d7e600b0d466d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-21 14:09:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ff6c7cfbb64e431d99536a57168acf5a : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e1faad3f78a7a878c16ca66b6586851024feec0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 15:49:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cf798b5b9824682a58910bc848e2716 : SUCCESS in 1h 41m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b3d4780d8ee74c24ca11ad0c3ec0d6150fb1614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 16:28:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7a544b8ff6bd421d80ecdd899759a530 : SUCCESS in 31m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c11f710bfd349b68ae60a93e9f215f9 : SUCCESS in 39m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e866a7aade1a4ef7a901b62744bf7196 : SUCCESS in 1h 31m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d73645b0765b4cfe98e2d5c263c284a6 : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/56f1859a28734cad9c05ba63f0f408f7 : SUCCESS in 1h 31m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c3f02471724f4a89b0ba404e8f025b95 : SUCCESS in 1h 01m 21s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/59605d4be641416487f1c0e333803e8e : SUCCESS in 2h 07m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3e03360ea5fc470e9e2997cd5a0006af : SUCCESS in 2h 22m 00s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b112562b27f349c284d0832310f9304c : SUCCESS in 1h 41m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86841952695945509d27eef400fa0ca3 : SUCCESS in 28m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62f2a1e27d6b4a61a4474e4b09836f56 : SUCCESS in 35m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5a3137b33e454d199f1e246bcce829ae : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0951b8a28f4478a869da8475442605 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901e6d4321a34434a3c803f195c69b78 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a060374a2004c54805a6ccf9f0fedb1 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9f166f844354e94a95e0a846a13f046 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5976673a1e314116b84d1f5199c67aa7 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4560178ce55e4de7add3144c63f3b59a : SUCCESS in 30m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc501504078941689d2152578abeebbf : SUCCESS in 42m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b62c04429c82466cb47a4212d8255a0a : SUCCESS in 51m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7322ceae510043649d4253b48d03c71c : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfdba7944a9d45e9bd6be0a7d7f6f4de : SUCCESS in 58m 35s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2ba74b45fb324319af8fed4b1a15c29f : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d47e8fbfa44b4369914c34d2197abb79 : SUCCESS in 1h 42m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b794a2084cbc432aba7552f386dca5e3 : SUCCESS in 1h 34m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/5983430b8acf44188c13bf14862d8c6c : SUCCESS in 1h 22m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a56c025d1d100df80037573a95d91a50d64ce9a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-02 14:11:18.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b02a2bd269543e309a36c95f9f3c8150ac531171","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-07 09:32:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ea218c69db5e1a3b5a12d564d6e91a97f80af118","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-07 09:40:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/2feb7f3db508488e828a963ee198ab56 : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"484ed374fe35588e05739991adcb71d9650ed84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 11:30:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/92f9b164374b46bebada6ec31fa02f3c : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0d467a3a7d434b16b4edca4caf1a2d25 : FAILURE in 1h 15m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c0b4f5e1b7c14e0abd0d542fc44d8013 : FAILURE in 44m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/50a2b9ad23644057bf583e4ea35e6826 : SUCCESS in 1h 29m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d635b0a65629491891f798a48a0b3dd7 : FAILURE in 32m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3200fab2368145bf9bd83bda79ee11f1 : SUCCESS in 43m 47s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/be05eedf320e40aa8b10e047bb0b32e1 : FAILURE in 31m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/972347f03e8f4eb9b47e915decd72087 : FAILURE in 30m 40s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/33997d7e1a724a4a90797ab70d59740e : FAILURE in 33m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c86545ce9edf400b9127e80d16d2e092 : SUCCESS in 25m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45e6dd24d6ad4e08ad890388b2530306 : SUCCESS in 22m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/63ac9386f253424da7b1c85e23f6fc02 : SUCCESS in 56m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e8fc84f5cc04ce2b121412035f1a0b9 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334ae9a596644b2782ae16905cafe464 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc6d451cb1dc4d038047015ed5247d27 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/702d344cfcfa4da5b1956df1dbc20f37 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce93cce13204989a53b5702fb1b5c22 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9880ec4d4acf4be58df1d1055cd3d77c : SUCCESS in 26m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf349c9b1de2434f982b6a0cb20ae2b7 : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4242ed349a984dbf9024315a2805c7d5 : SUCCESS in 1h 02m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/104ed309c5be4af6997d6b2752938813 : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ab3191a9f4149258da2f48efe0cf66c : SUCCESS in 1h 20m 42s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4a3c4d7b285f4146ae6eb4a9906d85eb : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/6c20d65d851e4b4ea95392381f05308f : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/436a87ed1036416c897d8e04f20c2e98 : SUCCESS in 1h 48m 15s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f07f34a37da34f3587d81fd238040b3c : SUCCESS in 1h 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f93102d347769b79c6012293af264e7e404d7f9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 11:38:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/831fda1e674d458e8a3c979d93a11247 : SUCCESS in 1h 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e64167e4f71568f96803422ec23ea078ace47ac8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-08 09:35:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d9ce51efb66d36788e2304dbb87ed2603a8c3318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 09:35:40.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"4e7e229c8f1db4860aa5ef4de78e4ba6e2fb85bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 09:35:41.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"c26cec69677e668e854199ebaa13af1b6c68bebe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-08 09:35:42.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"aff89a09b55aadd61b2c642deb074f468c4646d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 11:37:48.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"7b5b839321dd55ecb3702fd83feb4809071b1ad3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 15:00:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca92157826bd14f037870330b37d7b8817fdb732","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 15:12:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/86cdb586448742a380603d5c050741c2 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":5},{"id":"b377dda6fe57595cfb96e436364fe6f14ba57b46","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-09 15:41:47.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"08c669c2a05196ad82a2e2123ce54d90d15a62f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 16:55:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3b68718ed03fb3b93c0b006f2a2f5063ce7e5cb2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 17:08:12.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/649efbd81c064fbca3e40ffb8a2a253a : FAILURE in 10m 04s","accounts_in_message":[],"_revision_number":6},{"id":"78db731307d3c4ca6125d70aa238efac7e1ee772","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 18:02:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cb08d87f79d2042fde4dbcc8360215a5c1def5d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 18:15:19.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/455b607e61d1432f862c2e8a5ef0f79e : FAILURE in 11m 01s","accounts_in_message":[],"_revision_number":7},{"id":"f1193e6470e5712074aac972ffd4f31cc4a57f2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 19:39:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"79ceaa78603dac5864da3c0debaf30e853a99ca9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 19:48:45.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/bfc236b363a3487c907f10631699bb61 : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":8},{"id":"9a6a5c172da328924a56a708120722cc71db3a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:23:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8a531be876ff4b219c7ad463e69ff0ed : SUCCESS in 33m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6c804b3f98254feab473220d7dd2d100 : FAILURE in 38m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/36b677f3dcee445eb237be930c02d572 : SUCCESS in 43m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7bbd9e6217334c48ac0d3605945d6d3f : SUCCESS in 41m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2a495c7cf95244d98ee7473bd2e08d0d : SUCCESS in 1h 18m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3dc6ed0ba4134e8881d36d9d0e6d18f1 : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/24175ef3cf3e4af19a268b67e3a458df : SUCCESS in 1h 30m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/80bba3bad0c74eabab96e4e8471318e7 : SUCCESS in 44m 26s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/dc0caf16a26244dda300f22e9a4c5ee3 : SUCCESS in 1h 53m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/704c98b512a64257bc3ebf0994ed5563 : SUCCESS in 2h 17m 07s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/84b17b56a92440918937d4af8330bb1b : SUCCESS in 1h 42m 35s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/8799598846f145b1a400e6c921f0fd3b : SUCCESS in 1h 29m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/361854fc6f5c4067a3f58361e9eabeeb : SUCCESS in 28m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/08b77244bbdd447a9ebbc036edd5a1c8 : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/082bc6eedb5f4148be665cdd634f3da3 : SUCCESS in 1h 02m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6577f9f30cfc4bebbedc3cedbd15cb33 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e15d01d43b4309825da7949c1d605e : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e44e8e073c9478ea462d69ab4c6f2ee : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97bdf37d7afc428a95754d138ebb7a40 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8110ae84ec37422eb92226eae6ded344 : SUCCESS in 11m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cd2836447a6422ab9dcda90baed7574 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be877fcb30a944b3862474b7e4987689 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad1e2147e048442585e257429e14d794 : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f313c1b73c3c485d8ba1c4f174e879fc : SUCCESS in 1h 14m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ecadd7398f945daadb0d4c6adb7be37 : SUCCESS in 1h 09m 25s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/53d6af7ac8a1481d878067e51312dc97 : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8dda54df46cb4fffac5d66a7b88de2db : SUCCESS in 1h 19m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/cf7e2790680c4b23b558ca5ab5039ea8 : SUCCESS in 1h 33m 00s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e56c253136fc428da8b4e41fdef7fae2 : SUCCESS in 1h 14m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/27a3264092de438389475110e2672362 : SUCCESS in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bd8a2f3df67179790921ebe2197a015d9c05c73d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:52:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a826184b811147789842fa86b14d34bc : SUCCESS in 2h 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"44d2740a5b36232f3cc99c3db99a52a5af21046b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 10:34:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0c5b3269de5cf6234aad636ae2b58b7acda15f23","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 10:35:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c9722d208f3b940ec6f2e99970940f4fc6b047e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 10:35:49.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"42437a38300f5a190fa7ab80d18bb6d826516881","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 14:11:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2feb9b00632d831ced5733a1461c1ed07a41e383","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-10 14:23:44.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/b8824cda432a4872bf91580689187a41 : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":9},{"id":"b5bfc40d5a97191f6c622fd10b64f3d46c817673","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 15:46:50.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"8d8eb1f07eb6fac49ce5d3c7c576261827d1cd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 16:45:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/22df9cb6d01b455e94f2ef32fbc8d1c3 : SUCCESS in 34m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/412fed92444840d5980a4e912291842c : SUCCESS in 42m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/99a0279b3390438d8150d4bb0cc7d193 : SUCCESS in 47m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6fee4783591e4fc4b507ec83421cb41c : SUCCESS in 31m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/713186bdcd5549689e79555c9ff5c194 : SUCCESS in 1h 38m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/49b56b00bf32407480685715823d18ff : FAILURE in 52m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aa3ca690697040379b4cc3d3b6be4082 : SUCCESS in 1h 54m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3f3c5d6fdfe24cac89fdb461f6b8bd5e : SUCCESS in 56m 11s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0a2494548dad4c1985328ac6932da0c8 : SUCCESS in 2h 24m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7ef7c9d2e10c4befbe0b20712e363893 : SUCCESS in 2h 29m 04s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/578d2a7fcbe34666808145562085e200 : SUCCESS in 1h 42m 45s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7c65044554e847249890644c0f9d2453 : SUCCESS in 1h 51m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e65afd61432c4967937405c9f8027ecf : SUCCESS in 31m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce0edf06451b4f0c8ea6f70a59024aff : SUCCESS in 19m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f8ba95eecbb7488a82e90756400f9857 : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4872c4f6893343c999b82254f3a54117 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623e02dac649423eb3e3346fa4a6f823 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8f988a8d7541e5ba17fb5e28647717 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edf1ad3949234a5bab7a1ee878cd81e6 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44101169081943f2b27b363cb2044258 : SUCCESS in 6m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8096d05ac994bb3afc16747ae64fe12 : SUCCESS in 31m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b21b051c99c41589466165700ecba41 : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a34dcfcfc40c41c28428d7010747617e : SUCCESS in 1h 02m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ab3e3fcf9e7412fae72445d6c77da68 : SUCCESS in 55m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b9800605d21418989009735b811b9c9 : SUCCESS in 1h 06m 08s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7640ed2b73294d248539731c8bf16bf5 : SUCCESS in 1h 21m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/7c580835d5fd454c8d053d06ccd4e4f1 : SUCCESS in 1h 24m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f5a9c4ff35d84eceb0614bf280895b05 : SUCCESS in 1h 52m 05s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3d5025fa3fcf44fba873d3f58ed25696 : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8fd886800a344b659872683faea7b9eb : SUCCESS in 59m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"78bb19868a610728951ebb4e24ca564c02d15a68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 18:58:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/faea130a297e49429c465a0c6191382d : TIMED_OUT in 3h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"76c547c9e8ce5a51fda71a20078df465ecc13250","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-01-20 19:41:54.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\ntested this patch to build centos8 containers","accounts_in_message":[],"_revision_number":9},{"id":"b6afe28360887231500bb090375a1b92456b0a0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 13:54:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"65493de740865d293957de1e5ee2422fe2af92af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 13:54:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b3c2f9f7dda9bcd0ce0af992bb7d7d3bed0a03a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 14:14:24.000000000","message":"Patch Set 9: -Workflow Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"dc415d83e22f2748dbdb0761fcdd4a352c37dba5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-28 14:33:13.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"225582434eca39c341bbb06d5f5926fc6dca5337","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-28 14:35:58.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"fdf229eb877418da36bc939dae3b0b7ad84e18c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:50:21.000000000","message":"Patch Set 9: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"549781ce77e79d4a234ab7aa960b348c702c4965","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 09:47:06.000000000","message":"Patch Set 9: Workflow-1\n\nHold off until rabbitmq GPG key import issue is fixed.","accounts_in_message":[],"_revision_number":9},{"id":"38043e6b6ba47d93d639c7e8ef5e87e40a05c284","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 11:42:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"78f82bb3893361858a5963d0fb8a5d07d4a37652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 12:36:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f73aaf5327f6400294ff061309a02c67 : RETRY_LIMIT in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a672dec15968ed2c584008cc70b0773acc74a425","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 13:07:48.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/bf09d915ee8c48fa965e0374d0851d98 : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":10},{"id":"416c59b7fb8857aa08f03392ff0a9f5077e8f88f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 14:21:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/519c999738c04b42899751f321784b59 : SUCCESS in 40m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e3ebd124b8894f58bb4addd2db622cf9 : SUCCESS in 39m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8db21a16dbf341e68f706326b7562bd5 : SUCCESS in 31m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c0169b3213bb45e0a6f695b375490457 : SUCCESS in 37m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/02ea6855f64c4e789bceac6ed2327a01 : SUCCESS in 1h 25m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b94fe25d20814e16a9aa084c55b86250 : SUCCESS in 1h 12m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/868a4ab1b98d4d77b9a954b75d11f2b4 : SUCCESS in 1h 47m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d5ee0851a5784b01a54b315011559d6b : SUCCESS in 53m 32s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b87fb1cd30934bebbefe74d0c9bc30bc : SUCCESS in 1h 48m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/26084e7537024af990a1664d5307bb33 : TIMED_OUT in 2h 33m 50s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/dd3f6f24953842ada9db5625fc32d4ae : SUCCESS in 1h 44m 52s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3698b43b7a98449da4bf200d5a744843 : SUCCESS in 2h 00m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c131b213f8fa4eb686aa10acbe15aeac : SUCCESS in 28m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b01148d25c24449d934fe365004df624 : SUCCESS in 32m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c777aa00eace4a2bb1939cb019dfef71 : SUCCESS in 1h 35m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f12f544be5254eaf9886abb974ae0fcd : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75e45efd0e14a97a90df10bb6952314 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5fe7c63aa444d54acd38b29882a11ed : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/682ca8a83d0b4eac93c2c5986261bb3d : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765c5b110dab478db7701cf49fdbb983 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d974ab0d0e4c79bcfde2a4fbf2f5b7 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6865280d503c43a89536f98b0db7b3e8 : SUCCESS in 22m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49efc98ba21040dbb957dc0f0912ffef : SUCCESS in 38m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70e25583c7af439db3c7d9998b0adfdd : SUCCESS in 58m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9929a131c3794abaa8296d1a625b494d : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc9684fd90324ab8865fe0f68d2062e8 : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/af2a7e6634424a1eb370799123d42403 : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/5f4dd23dd163419bae72a0220aec9573 : SUCCESS in 1h 20m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/cd5eefe4e19f469088d6aa6887073d7c : SUCCESS in 1h 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e0dec8fbbb754aebb7558142960a154e : SUCCESS in 1h 14m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4b6e865f015d486a886a1c49563a3e34 : SUCCESS in 1h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f86ef2685fc014501175d74c8691d2fd7f156fbf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:22:11.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"6645f8b5f7b46a6390affcf5102e27e6a4bc8ba1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:22:30.000000000","message":"Patch Set 10:\n\nBased on https://review.opendev.org/#/c/704782/ which fixes rabbitmq issue.","accounts_in_message":[],"_revision_number":10},{"id":"59f9208d28f66d3a5d03ecce60b29ce347280e5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-29 19:09:59.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"f15f3ef7f7b84be1900f2e21993a2c13dc5f2945","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-29 19:47:37.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nInside comments are irrelevant; I want to know about this -ev drop - we want to remove it permanently or switch back to it when ready again?","accounts_in_message":[],"_revision_number":10},{"id":"cde69409551cea67203dfd018b93b3929974755a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 10:20:08.000000000","message":"Patch Set 10:\n\n\u003e (2 comments)\n \u003e \n \u003e Inside comments are irrelevant; I want to know about this -ev drop\n \u003e - we want to remove it permanently or switch back to it when ready\n \u003e again?\n\nIIRC, we used the centos-release-qemu-ev repo to get a more up to date qemu than CentOS 7 provides (x86_64 10:2.12.0-33.1.el7 from the last build [1]).\n\nIn CentOS 8 we get qemu from the advanced virtualisation repo, and it is more recent than EV (x86_6415:4.1.0-16.el8 [2]).\n\n[1] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_e3e/699414/10/check/kolla-build-centos-source/e3ebd12/kolla/build/nova-libvirt.log\n[2] https://dd2708c069c2d2c6f740-5c1e119604e6c3a613471c5b91ef3887.ssl.cf2.rackcdn.com/704337/7/check/kolla-build-centos8-source/2bbb871/kolla/build/nova-libvirt.log","accounts_in_message":[],"_revision_number":10},{"id":"80c23f7fc8c35311d06de4040e261ce8bea06563","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 10:21:51.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5fa911eab298032b32b777be59958de5a3821c19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 10:21:56.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nTYVM","accounts_in_message":[],"_revision_number":10},{"id":"d0a306e195315cd1a6da82d465c42ccc60371d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:22:11.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dacecdf7b7470b641d831fe6e5f80782bf08ca68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:27:34.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/30c1c05f167d4516ba29752bf8c0b6bd : SUCCESS in 53m 32s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/89ab97ecefcc422fb09b15975719756e : FAILURE in 35m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/40667a9f72be4ea894c9c6078179dd5f : SUCCESS in 1h 38m 47s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ab02e736e8024e328ce0a70e2b987b29 : SUCCESS in 1h 56m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/98110bf4ff1d47e5b4e3070dc51519c9 : SUCCESS in 59m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41dec6abdf9e462a8a813ba38092aee9 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593ddea107e0480db8a960076e523270 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6539487929264059a3510a322b67e20e : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dcfc34676814d0183e5afe1ce8e7001 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c90953e4dc402282d2f9029f4d4368 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bec820cc7df46dbb04a65c13db28aab : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95ea8e6c4639497aad0db09ab63d7ab1 : SUCCESS in 1h 19m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2bc5f67feb4c4a7c988c65c6ae31153c : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":10},{"id":"7a4e89be593ee081146e7b7589bfe91b189b0a9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-30 12:46:59.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"46fc64133b061eca66b4c889d25e0c96ea9fc507","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:56:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e6796d20d770b10a50a18039543d73b8d952a75d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 13:02:58.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/56fd33db97d54ceda4efdce157ff14c3 : POST_FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":10},{"id":"f1b50c9ebf8cd03e02c22e7a55ad6dfdf2a799dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:21:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/50993c1992784cf79741fd46c87a5f80 : SUCCESS in 29m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9e459974c0c94dbbb973cb6f70e49384 : SUCCESS in 45m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/995ae7d3cf8249ed840580f5195e08f1 : SUCCESS in 31m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ff1c9eba555d454bad4fb31c468f6621 : SUCCESS in 28m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1a1f058ab115461788edb9dec5d84816 : SUCCESS in 1h 20m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8a867d44db204a6cb1be7330c656f140 : SUCCESS in 1h 37m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/138d3752b45e49dbab48e2d78bbbecdb : FAILURE in 1h 03m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1f2714c9885f44578731eac6f91de61f : SUCCESS in 1h 00m 30s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/71915ce08e59403e9fd84e7e7bb01e3a : SUCCESS in 2h 10m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7bed39ac316f47a996db60eda7df56d8 : SUCCESS in 2h 30m 31s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/be498a3ec79c400ea3b43697a283eef8 : SUCCESS in 1h 51m 21s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/dce03303416f4865a3706c1845f67f12 : SUCCESS in 1h 28m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4cca588df05b4f5fa8963cad1c4b29e1 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1adc4dcfac64e9ea22240a4a1173e70 : SUCCESS in 25m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43c8849ee10a4040aacd69961f85e93b : SUCCESS in 1h 04m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f255b90beae46fbb1bc5c1c92889d0c : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4a176c80cc46bb9d79f398a3c2774a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338cf25d4619488bb226416aca79f2bf : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4268037e69654a9ba6c66497c18342c2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c291e8b3a18a4625ae3d0682414ba9a8 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7293c393900741bca71f8a86a8776888 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4d720e54c6044cdba69a596004e02d5 : SUCCESS in 34m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10dc3297ddfd497b91587bf1b90b27f1 : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7915a272f6a246f1a29ba387dc5cfa81 : SUCCESS in 1h 11m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8042a73c67146b4bee59fadc8f7f7e3 : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03b989cc8fa244a994c5c43997adca70 : SUCCESS in 1h 03m 18s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6f489b19f46548478e02c249eab1463c : SUCCESS in 1h 24m 00s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0cda22fc086049ae9ff545e4661319cc : FAILURE in 1h 36m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/abed95d2b1ac4cd9b1ceb213357efa90 : SUCCESS in 1h 44m 42s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/54fcb5d48adb4a7a8c3cc99e4837e186 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6807cc0a089b4554b3012075b6b10c84 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4b33f6d51ab78c8f4ea1ef4473aee900bd7c1852","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 15:53:08.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"34cf6387c586e2631451d59cf401c4812c171540","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 16:01:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"db419d7d628cd57e17ccd992dddd4e7e03aeb436","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 17:35:50.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7fdf80d2088840e59a1b8bddeaeaf683 : FAILURE in 6m 36s","accounts_in_message":[],"_revision_number":10},{"id":"310db540a70230ae01c5ae909afd45a3b380a252","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 17:49:42.000000000","message":"Patch Set 10:\n\nupgrade failed at deploy due to keepalived losing vip","accounts_in_message":[],"_revision_number":10},{"id":"86367f929069bc9ccb7d9db3ce662aa391edf190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:43:45.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6cd6756439174b618bc9b13b027d18f8 : SUCCESS in 32m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d6dfea9dd87447e4b968dfae3c2c9985 : SUCCESS in 41m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/38f8847bef8840bbb49023d556e6316e : SUCCESS in 24m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/03ecdfe300484ab08524baf1522c8bd4 : SUCCESS in 26m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/04928d396d044ffdbeb4778a26cf87b6 : SUCCESS in 1h 35m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/19ba687be86e4cdeba820b617d14a9de : SUCCESS in 1h 32m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/70ad99b6b283459589bfa782b4d8ebde : SUCCESS in 1h 28m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07566657dac6449db66d5e8545a4ae7a : SUCCESS in 54m 58s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f179c692e68c46788ea812790f57ca1d : SUCCESS in 1h 53m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f6b1ce53c92048d0aa2b306a53f72470 : TIMED_OUT in 2h 33m 33s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/43b88d6699ca4cb4aeb10357b8ef5a08 : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/36d4150f28e34a8686f91df4aee860eb : SUCCESS in 1h 34m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/872cf9b5ae3449b5bb3e15ad3781a3c1 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2914187f65764e06ba7e6045f26143b5 : SUCCESS in 28m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d49f49a9c0444355b24428a0ed3d5ab4 : SUCCESS in 1h 08m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2caa5dfe240a4d59862df353fbff6651 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102f1580406142fbbb62c5f2f02f1484 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2cc91e576f432084befedcb8889a94 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c24f68c87b2471f9cee050d9e42fc44 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/873049efeb5840b5accaa87603cb419c : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae87fe09aba4ce481488f934a9afa5c : SUCCESS in 9m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4761144b26f94f49ad06e66e10330e07 : SUCCESS in 27m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10d0a4e64ff44629921b56ff6540b542 : SUCCESS in 28m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86e2e0b7ae53406e8393d89db2e7ff62 : SUCCESS in 1h 03m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de948f043e8542fb80282ba3df9b1c26 : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d43b81c985e4dc6ba41145ab56045ef : SUCCESS in 1h 11m 30s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6d9ca4984fca4dbcb44417d9865c9ce6 : SUCCESS in 1h 14m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e93379abd11b48a1b172004d2663c1c0 : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4857b408ba764270a885c60db6a14cd8 : SUCCESS in 2h 16m 23s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/69ca77307e864ec7bc383dbfbbac0db6 : SUCCESS in 1h 20m 00s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/af9b53416bb04897a63e6ac184e08ebc : SUCCESS in 1h 22m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"70e399eef0965b4547a239ebe81b7e1e00b9ac18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:43:58.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4f7f9cee35d15d7e582263fea4fdc5389c3b3264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:13:03.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d95ef2ec8fd945a78bc0c34fab4d8262 : SUCCESS in 38m 22s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a3773b95eaa24ebfbb70d64ef7d3ae42 : SUCCESS in 23m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/35dcc3453aa14afd9a062e4979ae6105 : SUCCESS in 1h 19m 37s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b9614e9c5bfe492ba1cfba36a94c5a44 : SUCCESS in 1h 27m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/670ba13ebaf04653992f2b866eaed08e : SUCCESS in 49m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ccd9ac30ae54af5a55868f68db56b60 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b271bba2f78a4d759dc09512a19771e7 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/282c450aed78459b8266b56706cfc801 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b7831a9bf5943fd957db1fee7b8c95b : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33b14de1582433aa856917bbf990066 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d30f2eef7e994e68aed004a5fdc08435 : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a001767c5bbc49d5bf3393f1fedc2622 : SUCCESS in 1h 13m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d28f79ae644f4f60ae565ffa89a4faca : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":10},{"id":"25aea6855f87ae40d9456e09fa100b3d4e0d50d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:13:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"a1fea459aeb3d80eed5f72e9147b865c64025e88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:15:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e04a025f7dc418288d4d041e1446666 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4b4e26262b9e03190c639481fb60a8f4f92c532a","revisions":{"29cfdffaac0b2e4d5eaa82a36bd05e84214d7f25":{"kind":"REWORK","_number":1,"created":"2019-12-17 13:44:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/1"}}},"commit":{"parents":[{"commit":"cf5d7d4000cc27f524a176a44ebcc76b38493394","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf5d7d4000cc27f524a176a44ebcc76b38493394"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:20:44.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29cfdffaac0b2e4d5eaa82a36bd05e84214d7f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29cfdffaac0b2e4d5eaa82a36bd05e84214d7f25"}]},"branch":"refs/heads/master"},"d125951c9006b252ed64afba4ace64fe6e1212cc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-21 14:03:05.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/699414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/2"}}},"commit":{"parents":[{"commit":"e5f1e70b834a060aaf4eb0be6f43430b7297b2de","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5f1e70b834a060aaf4eb0be6f43430b7297b2de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-21 13:36:12.000000000","tz":60},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d125951c9006b252ed64afba4ace64fe6e1212cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d125951c9006b252ed64afba4ace64fe6e1212cc"}]},"branch":"refs/heads/master"},"11ea6ec9c5cde2d608fb3ac8e343018b1198b6ba":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-07 09:32:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/3"}}},"commit":{"parents":[{"commit":"3c1f5a2c665c5704088ccc28322d4654003998b1","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c1f5a2c665c5704088ccc28322d4654003998b1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-06 17:55:45.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11ea6ec9c5cde2d608fb3ac8e343018b1198b6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11ea6ec9c5cde2d608fb3ac8e343018b1198b6ba"}]},"branch":"refs/heads/master"},"fe839c8cc6dd77fc9a215407bbf717ac1c38a08b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-08 09:35:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/699414/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/4"}}},"commit":{"parents":[{"commit":"5f7552162c0eeaec32a5a2e983c823fee7281ac3","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f7552162c0eeaec32a5a2e983c823fee7281ac3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-08 09:35:27.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe839c8cc6dd77fc9a215407bbf717ac1c38a08b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe839c8cc6dd77fc9a215407bbf717ac1c38a08b"}]},"branch":"refs/heads/master"},"a53cbe66a82f69aba1c0d75a19265e7d1c8133d7":{"kind":"REWORK","_number":5,"created":"2020-01-09 15:00:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/5"}}},"commit":{"parents":[{"commit":"a0b676337409bbe99cabdfc0e8a38e8133316547","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0b676337409bbe99cabdfc0e8a38e8133316547"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 15:00:05.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53cbe66a82f69aba1c0d75a19265e7d1c8133d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53cbe66a82f69aba1c0d75a19265e7d1c8133d7"}]},"branch":"refs/heads/master"},"b809beb8ed70313b05ff104105d5c4417d65e4c9":{"kind":"REWORK","_number":6,"created":"2020-01-09 16:55:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/6"}}},"commit":{"parents":[{"commit":"7c6f700f7a2c86f95aed55752dde6c7557e80bf6","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c6f700f7a2c86f95aed55752dde6c7557e80bf6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 16:32:59.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b809beb8ed70313b05ff104105d5c4417d65e4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b809beb8ed70313b05ff104105d5c4417d65e4c9"}]},"branch":"refs/heads/master"},"e58f64e9857bbda91e75e3fb94a85110e23bdb8c":{"kind":"REWORK","_number":7,"created":"2020-01-09 18:02:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/7"}}},"commit":{"parents":[{"commit":"3c618a1f6f0beca723e031fbe323b159400ad08f","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c618a1f6f0beca723e031fbe323b159400ad08f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 18:02:30.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58f64e9857bbda91e75e3fb94a85110e23bdb8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58f64e9857bbda91e75e3fb94a85110e23bdb8c"}]},"branch":"refs/heads/master"},"7680a6552fb4f5278aa2e11dac07ddb3f14c56b7":{"kind":"REWORK","_number":8,"created":"2020-01-09 19:39:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/8"}}},"commit":{"parents":[{"commit":"3c618a1f6f0beca723e031fbe323b159400ad08f","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c618a1f6f0beca723e031fbe323b159400ad08f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 19:33:43.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Don\u0027t install scsi-target-utils on CentOS 8, it is no longer available\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python-3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7680a6552fb4f5278aa2e11dac07ddb3f14c56b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7680a6552fb4f5278aa2e11dac07ddb3f14c56b7"}]},"branch":"refs/heads/master"},"d1304acb318454ddf0410d3f41d2c4f2d00b2c3d":{"kind":"REWORK","_number":9,"created":"2020-01-10 14:11:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/9"}}},"commit":{"parents":[{"commit":"3c618a1f6f0beca723e031fbe323b159400ad08f","subject":"CentOS 8: base and openstack-base images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c618a1f6f0beca723e031fbe323b159400ad08f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-10 12:27:53.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Switch from qemu-img-ev to qemu-img on CentOS 8\n* bridge-utils no longer available on CentOS 8\n* libvirt-daemon-driver-lxc no longer available on CentOS 8\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python-3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1304acb318454ddf0410d3f41d2c4f2d00b2c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1304acb318454ddf0410d3f41d2c4f2d00b2c3d"}]},"branch":"refs/heads/master"},"4b4e26262b9e03190c639481fb60a8f4f92c532a":{"kind":"REWORK","_number":10,"created":"2020-01-29 11:42:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/699414/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/699414/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/699414/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/699414/10"}}},"commit":{"parents":[{"commit":"bdb0dbddbc30215169baaaa263c583e7d3183607","subject":"CentOS 8: base image fixes for RabbitMQ, fluentd \u0026 InfluxDB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdb0dbddbc30215169baaaa263c583e7d3183607"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-17 13:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-29 11:41:38.000000000","tz":0},"subject":"CentOS 8: Update packages in images","message":"CentOS 8: Update packages in images\n\n* Some further changes for python2 vs python3 packages\n* Allow rabbitmq 3.7.*, since a newer erlang is available\n* Switch from qemu-img-ev to qemu-img on CentOS 8\n* bridge-utils no longer available on CentOS 8\n* libvirt-daemon-driver-lxc no longer available on CentOS 8\n* Mark some more images buildable for CentOS 8\n\nChange-Id: Iaf5b68ff6d944ae730ca0b1d5832172c106a6c08\nPartially-Implements: blueprint centos-rhel-8\nPartially-Implements: blueprint centos-rhel-python-3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b4e26262b9e03190c639481fb60a8f4f92c532a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b4e26262b9e03190c639481fb60a8f4f92c532a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
