)]}'
{"id":"openstack%2Fkolla~699483","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1856699-stable/train-stable/stein","hashtags":[],"change_id":"Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff","subject":"Fix monasca-grafana being unbuildable due to old npm","status":"MERGED","created":"2019-12-17 19:19:53.000000000","updated":"2019-12-18 14:57:34.000000000","submitted":"2019-12-18 14:55:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699483-1576680933283-bbef5f2b","meta_rev_id":"a89387da1a1c2151540500ebd546bc56562fdad3","_number":699483,"virtual_id_number":699483,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-12-18 12:19:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 14:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-12-17 19:20:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-12-18 09:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-17 19:24:40.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-12-18 09:10:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 19:20:13.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-12-17 19:24:40.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-18 09:10:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-18 14:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ba089dc17d66d1685f7b4ee677ddf781286055f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 19:19:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"bc394e10df1898f7eb89f5da498e31eedbfbd971","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-17 19:20:02.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit f53095af3feaa47419719bd1e7542464d68b5fa3","accounts_in_message":[],"_revision_number":1},{"id":"43298e58501ac39a0e6d7fc0a0a0769eeefd4cf9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 19:20:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"51785e7ac28a72cd3c84846407afd830fa45d4f1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 19:24:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5eaa1982ca6a1df012e09f9f630b59217090dcbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:14:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b9ffd71095c84d0d9cd9ea6df56fcf08 : SUCCESS in 54m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/76ee0eb73bc9495fbb3ebd1fb4b00ab0 : SUCCESS in 50m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/db64af298b5b46c2b620568f3e3bdb26 : SUCCESS in 1h 16m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2408d069b0e84a29b61cd24532ed167c : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/afc6fef96c774c6292a4bd7c5e97abc4 : SUCCESS in 1h 47m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/419ec5cf2d4d4e8ca5d09d1b363c47b1 : SUCCESS in 1h 02m 00s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/856398263c594548b368fc0231b020d6 : FAILURE in 44m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/455d17c3969747ef9e52da02158571b6 : SUCCESS in 53m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ced4f0c93271411eb5b635551be60e08 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83607c84720e4dafa6382cf7343b982b : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7b590b5353640e18d6e241be15f581f : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f33552290134042806a0a81cfe3e983 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7243816d2afc406e8035cbfe88476d40 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75e774cc2e14e978d1740277cfefda9 : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0cc16e4e6ee4411fa9978a4a23c86f35 : SUCCESS in 45m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ecd94bb644a4d8295c53922f09a4986 : SUCCESS in 37m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2407a9ff5dd849bdbd839e4508426920 : SUCCESS in 58m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cac3f3fb90742fc9263d0ebf3a23a61 : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fcfd933f2ad44e1e8931a80b92b76ec7 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f874cca7a34ca15a4925d8678d4699dd7d933e57","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-18 09:10:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d1b6d6744438a6b9d7ce690156c61cbf125be87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:10:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"770bff081f19b6407deb7665179dccb9b804075b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 11:15:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7765189890c6488590be92aca657f387 : FAILURE in 1h 09m 02s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a93675170640444f9faf2ed4b01928f0 : SUCCESS in 1h 08m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8102bab9a7eb4a1e85c50aa2585424f5 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81153810890a4b00827b38663b95e863 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d280f67bb124414a74e8acfa2c4f5f2 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8aeb87227f24c90817929c14d788a69 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bc8abf350c64eb68d0a16764483ed9f : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48566825e841478891f3a6019b4d8191 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9cde3dc8bee48ffb4f6c6899891a527 : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"64c48aff4627cd54f67693de16185fa7695191ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-18 12:19:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a05f0aa6686a1f2a824256fe059540d4ffa145c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:55:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/197c20a7f2d240bcb346b2d514ba9a6b : SUCCESS in 45m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4120464ab86c4ca88038cb4b751570a9 : SUCCESS in 52m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/82dc3690749941a6bc7051803cee48a5 : SUCCESS in 1h 09m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fdc7736890da41eba6e9543a2e43c6db : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/568927faadf64049b72b95ad0bb80793 : SUCCESS in 1h 30m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/08a29dbe447c4906bf76cace3afa216c : SUCCESS in 40m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3c4997d72016484eb5eaefa083a81d69 : FAILURE in 51m 24s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7a02a003a8a344b88d5b1952b21735e5 : SUCCESS in 45m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb04291c44ba4ed581bee58cefa786b5 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52808f913de04b41b93cd10c39a9f3ad : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1f6ae8b13444b9fb8898d68b81f7d79 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d70f0349a2274b648b938d27cd8d7e0e : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/376db4dff8974fe7b21f01fa4b609d4c : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6cb85ff6914121a4f5cfd910541767 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd0537e2eb1244f09c740adeb8e45640 : SUCCESS in 28m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76320cbf784b408b9d99848847ceac74 : SUCCESS in 30m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/371133d4a8a6448c9ac7c67851c73a0c : SUCCESS in 1h 03m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90ad51d87b4d4d128374f966e133758e : SUCCESS in 48m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/81cc492dd6b24031a1d53b8c1c2584ba : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"57915eda42c960179b7ca833e61c5192fa5de645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:55:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7c3c669dffd498304955a60d0d5b31d41ec9cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:55:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dafa077e8f3844efb094d41795736492 : SUCCESS in 51m 37s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c141dc631cee4366aa9284e5db481409 : SUCCESS in 58m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d85b9c908d054b97b36019c75263e3e5 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397574a72ecc452192310174cb5a3868 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96cf4d3fd7624c51a35c5037b92249c2 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/436c415da68547c587d4a6da3bf5dda7 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f67b4516068247cc9ba55cb29c39b5c7 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855cd07bb13b41ea89173b15f78aaf45 : SUCCESS in 3m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd840f818a6e46969575438d8ae45c96 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"519700fce4295680c371fe037b1aaef6db5df151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:55:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a156030648b22bd0858dd6c8e1fd7bdef72e9fc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:57:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f318f7841944568d381d354352da86 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acbc6e0eb95868adf58f1916669ecd850d049758","revisions":{"acbc6e0eb95868adf58f1916669ecd850d049758":{"kind":"REWORK","_number":1,"created":"2019-12-17 19:19:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/83/699483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/699483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/699483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/699483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/699483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/699483/1"}}},"commit":{"parents":[{"commit":"7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54","subject":"Fix keystone bootstrap error message display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7934cf7d4a18e90be38cb4c40f39b1dfd4f49a54"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 11:57:49.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-12-17 19:19:53.000000000","tz":0},"subject":"Fix monasca-grafana being unbuildable due to old npm","message":"Fix monasca-grafana being unbuildable due to old npm\n\nAffected builds on Ubuntu which had:\nnpm 3.5.2\nwhile the latest was:\nnpm 6.13.4\nAnd CentOS had:\nnpm 5.6.0\n\nThis patch runs npm update via npm.\nPinned to 6.x.\n\nAdditionally, this moves npm packages prefix (root) to\n/usr/local to avoid conflicts with native packages.\n\nChange-Id: Ibaacc1cc478b2b2f8196a5da4eea3570d7d310ff\nCloses-bug: #1856699\n(cherry picked from commit 308e85aa413da84ed107b27fa09a3f0e0a64392f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acbc6e0eb95868adf58f1916669ecd850d049758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acbc6e0eb95868adf58f1916669ecd850d049758"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
