)]}'
{"id":"openstack%2Fkolla~699497","triplet_id":"openstack%2Fkolla~master~Ie0b39743e2dbfaac55a6467ca98fb2f5a53a49e9","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie0b39743e2dbfaac55a6467ca98fb2f5a53a49e9","subject":"Add AArch64 publisher for Debian/source","status":"MERGED","created":"2019-12-17 20:48:25.000000000","updated":"2019-12-20 07:55:49.000000000","submitted":"2019-12-19 18:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699497-1576779709617-7a9322a4","meta_rev_id":"1b2c85b01f8b00f895a167c9a5898f463f20f921","_number":699497,"virtual_id_number":699497,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 18:21:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-12-19 08:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-19 08:33:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2019-12-19 08:51:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-12-19 16:39:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 09:10:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-19 08:51:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-19 16:39:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-19 18:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"068de1a040b8edadc8c71ef57c8d97d310d63f50","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-17 20:48:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f3b3557a5bd2bfb80bb9c29c78aaf484704f5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1754c9b7484d4cb68809643b3879348e : SUCCESS in 39m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8fc94e2f5c146889f37bc0fa3b37830 : SUCCESS in 42m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/21c3509c48c443d592bb5f5c9aa647b5 : SUCCESS in 27m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a8816a01074a4530919340eb88c0d7f5 : SUCCESS in 35m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7b7ec6c26a2462fa68196169b479cb1 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bee60369c04ba39aa011d36c5334b7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1559f8a44864abc8ea5a79849a3d75f : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8c9fb599a7446d9b76c4b23023de73b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39c3695609c49b7b408db004b44693e : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d13948b903f84296a8b1b0d72f968684 : SUCCESS in 25m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6c3f61e9a50417b9ab11c83c8adb27c : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"778ad2596de72b288f58adf5455246e8d2b939cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:55:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba9e1993bf464c1e9bbc45ce164557eb : SUCCESS in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2d1511b17e97c8adbe8eefe3c33bae7d348815c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-19 08:29:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"515fe70e41d2295907143552b7ee41808aef2a00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-19 08:32:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f2d10aa74a9d7216b14cf2a5c3dbf76c0878d26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-19 08:33:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b59863eea3e8ebb91b1c9ca67bbf70ab648d009","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-19 08:51:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a5e76b659157a81645c98e759aaf04d96f6b4290","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-19 08:59:04.000000000","message":"Removed Workflow+1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"e9acece9f6a251abe68d9ef8125e4941cbe9e677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:20:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/699c85c132c947d8b1a69ef2664cc317 : FAILURE in 34m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/271c564e83034ef0acb6a9f5b9892d0f : FAILURE in 39m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/92cfbf285de64f44bd85840ebbbcc79b : SUCCESS in 35m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/51267a3badaa4e83a3dd58a0316a3250 : SUCCESS in 17m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b75be49ce49b475f841b76c42c9965ce : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f981e1328d264dca8010497357861527 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8f012fb77794603a12b3e0f1169e52c : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba8804b213e843aba19658549a7e842a : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c835451d12440daace5cc2284538233 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6724d1043d8e4e619aa228692417d521 : SUCCESS in 30m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1dc364c518264444bb12dc2e697dc404 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b0bf529eefca693b84339c407f3011f6e88adb92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:44:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5fd27046a814b2eb8872f7536049275 : SUCCESS in 1h 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac64f46795312760c05995a8208f11f62c7691e2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-19 16:39:35.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"60aa8f7ee4d933ac0eee383995523adeb51df9d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:35:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f1fa99f7bb464d08b78ef7ac71c25c4c : SUCCESS in 41m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/17b52b788e32487c9483ae5694b0d4ee : SUCCESS in 45m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b588bef613884c66be979bcdcbb8d0e3 : SUCCESS in 23m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/53d1ef1b6d9b4bd7993cb71701a0d088 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea8226febfe4e7da5cdf2cdbb87fcc9 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdd92756a71451f8c5c33d53ff19e3f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d7185290f844672a3e30f836534151b : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a0a52576e094c9ba5933a439eb08e83 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd30496f0f704218b0422cd3df5a8249 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e87d08f600d4479b7267e273a1f1582 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b63c4684e68e4c26a191bb7727e2d2c5 : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":2},{"id":"89b7e81c3569bcfb34c04176e97967ade870f3c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:36:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd4684b462a29748fd1c398d91145ffb275445d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:45:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d046b7c7df2148edaaa13f77dd679543 : SUCCESS in 1h 03m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d98a513f88a1dec0e2a474e2f710a719e4956a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 18:21:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1ff7b9dd0b234296a2508b4a0cd9c30c : SUCCESS in 44m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8efe197c38ea42fea448ba6ab30c20cf : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b81f76ce6e485b854f2395c6520b41 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b0eba41f5bd4b5cb5dd363773fbdc62 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a4b7c2c8f65499a8db429d72cfd42b5 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4033fb1ecb43eeb8b0527a3f2d1894 : SUCCESS in 8m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ccb0a5245dd4dbaa577ebff267811a7 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d247869e2f20163357439aaf886bc2b1a9323257","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 18:21:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c363330bcf2d1ef3f382775bb5ad7bd0186cf3e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 18:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4860ea0326f4be49c27429b6140d641 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1b2c85b01f8b00f895a167c9a5898f463f20f921","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-20 07:55:49.000000000","message":"Created a revert of this change as I57e725b44efe16a9f2f3900c6e14eb84aac37c32","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61fed77d79c8242c64772df5d409fdaffdc3a150","revisions":{"73852d54d404ca927bb784ff2c21bbcd702c62ba":{"kind":"REWORK","_number":1,"created":"2019-12-17 20:48:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/97/699497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/699497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/699497/1"}}},"commit":{"parents":[{"commit":"308e85aa413da84ed107b27fa09a3f0e0a64392f","subject":"Fix monasca-grafana being unbuildable due to old npm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/308e85aa413da84ed107b27fa09a3f0e0a64392f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 20:47:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 20:47:14.000000000","tz":60},"subject":"Add AArch64 publisher for Debian/source","message":"Add AArch64 publisher for Debian/source\n\nWe want to have AArch64 in kolla-ansible CI. But building images each\ntime takes time. So let build them once, publish and re-use.\n\nChange-Id: Ie0b39743e2dbfaac55a6467ca98fb2f5a53a49e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73852d54d404ca927bb784ff2c21bbcd702c62ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73852d54d404ca927bb784ff2c21bbcd702c62ba"}]},"branch":"refs/heads/master"},"61fed77d79c8242c64772df5d409fdaffdc3a150":{"kind":"REWORK","_number":2,"created":"2019-12-19 08:32:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/97/699497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/699497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/699497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/699497/2"}}},"commit":{"parents":[{"commit":"308e85aa413da84ed107b27fa09a3f0e0a64392f","subject":"Fix monasca-grafana being unbuildable due to old npm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/308e85aa413da84ed107b27fa09a3f0e0a64392f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-17 20:47:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-12-19 08:32:10.000000000","tz":60},"subject":"Add AArch64 publisher for Debian/source","message":"Add AArch64 publisher for Debian/source\n\nWe want to have AArch64 in kolla-ansible CI. But building images each\ntime takes time. So let build them once, publish and re-use.\n\nChange-Id: Ie0b39743e2dbfaac55a6467ca98fb2f5a53a49e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61fed77d79c8242c64772df5d409fdaffdc3a150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61fed77d79c8242c64772df5d409fdaffdc3a150"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
