)]}'
{"id":"openstack%2Fkolla~700184","triplet_id":"openstack%2Fkolla~stable%2Fstein~I6f6b32e3858f793a4674967e3fa3fcec87d93e6e","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I6f6b32e3858f793a4674967e3fa3fcec87d93e6e","subject":"Bump OpenStack versions for Stein","status":"MERGED","created":"2019-12-20 15:01:06.000000000","updated":"2020-01-02 13:03:10.000000000","submitted":"2020-01-02 13:01:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700184-1577970086870-d3a5649f","meta_rev_id":"00511e0f36ed30c5594cdd006ab7bcd9f362a735","_number":700184,"virtual_id_number":700184,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 13:01:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-12-23 11:31:49.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-20 15:01:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2020-01-02 12:02:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 15:01:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-23 11:31:49.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-01-02 12:02:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-02 13:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eda13430a31e3be6ba5f27c6e314484a7de00823","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 15:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58507a512a0cf89123d4206d57050782bda699de","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-20 15:01:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2edb1ce155a4cbf9c780188bd8cee814dbec7ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:51:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c46781fdcb0e40d681f62fa633c1dfda : SUCCESS in 41m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/768e0cde392e4f3fb510d0e090154043 : SUCCESS in 48m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/992d4b2ee14540fc9e63d999d82796db : SUCCESS in 1h 23m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/911c77f9f66f4a629c82850cb0bf1e74 : SUCCESS in 1h 13m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bf6b059389944ea292d545be2f6e9b05 : SUCCESS in 1h 45m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fbc02a54e16a44cebc85b72c1893da4e : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/72907a7fc32043639a835518e2d283a6 : FAILURE in 42m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e6c7b323f97e43ca907f3359feb25786 : SUCCESS in 55m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/656005f254184b828d39b6ab9c6a385a : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cf6e37ce774ea7a82eaaebcb877be2 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09864c48e12541088271a549d41cce65 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40ce746760064abf96863d082f11b4b6 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9c64dcfdc6e4cbfba1a48961b989486 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b26257db3a48e59ac81d6c536cfd28 : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b59ac2c96f6478e9c88526f950d867b : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5a01a2830cc41ab8487cf279ee909fa : SUCCESS in 31m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1dc37bae3e8d47a2981e69e7745857b7 : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2503e8c4539e4d7fb0e95ee2acb48ae7 : SUCCESS in 54m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6dfd18dbc9fb49169815fb86105004af : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"59932ee988fe0dbae19615e34e1b723fe63dd69c","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-12-23 11:31:49.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cdd6ffd5372c18f2d4ec84a4d735aa8a663585f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 13:12:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7f0fa9065bf74775ae4e37aa831ae0d1 : SUCCESS in 53m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ca3b3bc7d0f141ff859210d29927557f : SUCCESS in 1h 04m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/20542513e71b47c8b95d18c0c5036c2f : SUCCESS in 1h 26m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/54823bb0724a493eb4e1630e1f1df4f5 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e10c44de46ae4e319046f9b2e60f3825 : SUCCESS in 1h 24m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3b209049bcdd4c26b214ba092c948834 : SUCCESS in 41m 24s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8a2fe4826e134ab0b987ce66dd8b8da3 : FAILURE in 44m 15s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/af72d7ad818c411585beb46501d3a3f9 : SUCCESS in 43m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/552e22faf98b43e4a4470e9598d72281 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f886d552462c4b16ba96b69ca02aad1f : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4febdc65974d4c7bb9f3b0807e1bb445 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11792c128c334d86916e01ff2c9ff60c : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66a21df557924947bb820a33ceeacba9 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dcd18d0d5e84d2b94cefb920d47e320 : SUCCESS in 3m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bfc88c68b60945e79abc070491793113 : SUCCESS in 24m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f910a114fc5641689ff3048fffe1e8b6 : SUCCESS in 29m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34099fc352eb4051a13318b6f8184da2 : SUCCESS in 57m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e86f25eb54c4f56bcd6485a434fd350 : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5b15b6502514f1c8c360fdf2861d082 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9c6e019a9ecc99ebffd404beecbdfbb3e0bb6723","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-02 12:02:57.000000000","message":"Patch Set 1: Workflow+1\n\nOnly tripleoclient image failing on Oracle binary.\n\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_8a2/700184/1/check/kolla-build-oraclelinux-binary/8a2fe48/kolla/build/000_FAILED_tripleoclient.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"f96c9177aecb7c5b629124bd73aee5e860eaf768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 12:03:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b93553c08cda5ab259758f5dcc09edfb156c73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:01:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b58f0977711cd94c7376acb319b1ce38b4ed23dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:01:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d4416c50af7e4ac6845dedbb2e9ede1f : SUCCESS in 48m 06s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/eeefe856f913470c82e095c2bd274c6c : SUCCESS in 54m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f901ed78aef645478268a3620a9301aa : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c91aabf52f43e79a8e9447f9e06e06 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/172fb97bd09e42d99989f20f23921d77 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ad91508b88f4d9e876ab6a8ca25c286 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4413584550043b7b122b49482f3a593 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe97c0b2fb5496999d745ccb6acfffa : SUCCESS in 3m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/958539aff7f14ce3be55ee91d2565b93 : SUCCESS in 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"00511e0f36ed30c5594cdd006ab7bcd9f362a735","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 13:03:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f5274d862c4eca8ab56d7e27861fd2 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56772b04eaeb901cc3e796974a79a512cfa4cbce","revisions":{"56772b04eaeb901cc3e796974a79a512cfa4cbce":{"kind":"REWORK","_number":1,"created":"2019-12-20 15:01:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/700184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/700184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/700184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/700184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/700184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/700184/1"}}},"commit":{"parents":[{"commit":"a7b0f25d4b21e013032c92e3160697f8d996480c","subject":"Bump versions for Openstack stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7b0f25d4b21e013032c92e3160697f8d996480c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-20 14:54:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-20 14:54:53.000000000","tz":60},"subject":"Bump OpenStack versions for Stein","message":"Bump OpenStack versions for Stein\n\nChange-Id: I6f6b32e3858f793a4674967e3fa3fcec87d93e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56772b04eaeb901cc3e796974a79a512cfa4cbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56772b04eaeb901cc3e796974a79a512cfa4cbce"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
