)]}'
{"id":"openstack%2Fkolla~700185","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I043e05b65622e9e9d27553c8519b40d6702964c4","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I043e05b65622e9e9d27553c8519b40d6702964c4","subject":"Bump OpenStack versions for Train","status":"MERGED","created":"2019-12-20 15:01:23.000000000","updated":"2019-12-23 13:28:03.000000000","submitted":"2019-12-23 13:26:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700185-1577107605325-17819b5d","meta_rev_id":"8483068aeba986100a614bec76a7b3118bf97a14","_number":700185,"virtual_id_number":700185,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-23 13:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-12-23 11:31:22.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-12-20 15:01:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-12-23 11:31:22.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 15:01:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-23 11:31:22.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-12-23 13:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f33014fa60441ffb048d8406ba01bccfc82f4ad8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 15:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c904046393e80b32dda1f89b6e39035ac4174bf0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-20 15:01:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a35544cdbac37dfabddfe6acf34c630c09ddb93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:51:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bd9c37df05bd4c2c847d20c8d4ffe053 : SUCCESS in 43m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0cdfb7b9fe3d49fbb510a78878538c6d : SUCCESS in 46m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8403a64bb40a4f72ad82a5c5b4a9e815 : SUCCESS in 1h 38m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cb1561b1485c4ada8971809154abb95a : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c020f4f609b64544b1959df10508c2a6 : SUCCESS in 1h 24m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d3364a90352c4ad397db7e8d4fb6625d : SUCCESS in 1h 02m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed9d69f8071244de9cac3fa612592b60 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6101ffe7819145d7b6e2beb10956b201 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6292d076b49448cb908dc9a1f100551c : SUCCESS in 1h 07m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aad3dbd2152349e2bce4f81e4e0e935e : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23173dd8e384bb0bf5f035fdb73c6e4 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54e118dd6d1f4cadaa0ec6c826b80806 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/362af20443454eeda12573e38745ab49 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/151f3363a9aa41c98b0fdb06c74ca9e1 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1069be8d7a884071a78e127703ad49d3 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34baafd0350b4af2a4903c5869adec2d : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4ad04568ae44008be55dcdce6efe813 : SUCCESS in 30m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/869616597ca34adf8f1a38824924544d : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/020061f44e434aa0bfb51f92e8ad4272 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ecf7bfcb16f41beb85b54d173ce4be0 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1b70ab1dd2f786d369b40ff753960714420c3d5e","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-12-23 11:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3e037df25186a678e3496333ee45b26e974c20d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 11:31:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ccd3803991675dd26561eaf12f34fee7a246c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 13:26:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eb6986f29e284528bc32e412105713bc : SUCCESS in 1h 10m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0a14d9444eb4428d80c5589d39a85da8 : SUCCESS in 1h 31m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ee63d39c72f4d00b822fbde23490350 : SUCCESS in 1h 54m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2b4ddaa3dbce4a14b47a52a5641e2eb8 : SUCCESS in 58m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/172a421f6da34e28bc1cc1300c8c916d : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25736f3d3e64876bbd8cf894656effa : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d74aaa0787554a4d9c5de2575009b617 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc7efb56e7b47e4b1ef88d0f5f77b25 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66f6b45582654495abb06164bed94a46 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d920555733145d59dbed8e073238517 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d16cbe51387b4f73a7a9d92e42d6dff7 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd7e098c4d8f489fba89697b15db67da : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8720079afbdc4aaf98e4e6ce2597007b : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"bd83bc77e554b5daef2fc3de1827778c031d9e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 13:26:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8483068aeba986100a614bec76a7b3118bf97a14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 13:28:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2854e296707a4dc38b2aebb572dcb863 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16422e1ef44239a581bbdd89ac8da7bbcd2bd5d8","revisions":{"16422e1ef44239a581bbdd89ac8da7bbcd2bd5d8":{"kind":"REWORK","_number":1,"created":"2019-12-20 15:01:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/700185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/700185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/700185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/700185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/700185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/700185/1"}}},"commit":{"parents":[{"commit":"582fe930005165f2d7345c3330b8c36135ddc556","subject":"Bump versions for Openstack train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/582fe930005165f2d7345c3330b8c36135ddc556"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-20 14:51:20.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-20 14:51:20.000000000","tz":60},"subject":"Bump OpenStack versions for Train","message":"Bump OpenStack versions for Train\n\nChange-Id: I043e05b65622e9e9d27553c8519b40d6702964c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16422e1ef44239a581bbdd89ac8da7bbcd2bd5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16422e1ef44239a581bbdd89ac8da7bbcd2bd5d8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
