)]}'
{"id":"openstack%2Fkolla~701121","triplet_id":"openstack%2Fkolla~master~I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5","subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","status":"MERGED","created":"2020-01-05 21:18:26.000000000","updated":"2020-04-09 14:57:11.000000000","submitted":"2020-04-09 14:50:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701121-1586443833066-30de0cd9","meta_rev_id":"ec34cf62130123bb0b957620c9c8a9bb3cacccfd","_number":701121,"virtual_id_number":701121,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 14:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-09 12:59:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-04-09 06:32:21.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-09 12:59:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-09 12:43:14.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 09:51:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-09 06:32:21.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-09 12:43:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-09 12:59:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-09 14:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfc0f0e22aae5c1b24406a4b380cf9821db845c2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-05 21:18:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf2ac36da88961df0b1bbb9a32fbc319f2d69f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 22:06:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cba752585e84425f84982f960a721186 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe6771a9196c48b3bbcffaaf116abe7c : FAILURE in 35m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e4f36edba383489b9dd1474cbbb180c4 : SUCCESS in 35m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f24dba387fbd4ad7a13cb7ff14f26061 : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52fc7bdcb34947bdaffd58a30569105f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4742345f7fb481c9ffeb5a98413d825 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd6c233fb45b4d249c8a0fcc061d6441 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25b3bcde29944bc7a7bacf6cf4cbed95 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335c1f7e2ac74876bec90638ce66a1ea : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bef31711656e4f40b61444e3b225dad4 : SUCCESS in 32m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05d14821218a47b584442b1255cc9b00 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"879cd143a6bde30b1347c562bb0a42d334835a43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 22:56:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/882acc52b55340efa63f3c10bb7eb4bd : SUCCESS in 1h 30m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/29029ef204724dc4afb075cf51b4c725 : FAILURE in 1h 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c9673ea8c668d4f549cf46f2d79a35190b6b189","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 09:51:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68f57c7ca1d4f9f6fade094132c088bb8c09fcd5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-06 13:40:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01c5e4c997306b444a9e2f1d9e8e2b90766bc498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 14:49:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8dacc20cb4c443189f11351679a5c838 : SUCCESS in 57m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f77a0e9c64554062aa13a9b7244fdb28 : FAILURE in 1h 01m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3f8d4284bc2449e9a814d77d7e87b81b : SUCCESS in 34m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6f4c9d4bcadd4118b419291a3b6231a4 : SUCCESS in 26m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/871f49983c56499e88872f60e1752a8e : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f222ea693f994d9cbd8d53e1619e489b : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed22869503eb47de937894c64a50c857 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c02e4bdf8ffe45df8beb57a6a947a4a5 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbcf63101a04607983c50c8bfcb1c95 : SUCCESS in 9m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5181dba818864ae3afd2a4896261fb88 : SUCCESS in 41m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b21dfba2474343df837a975cb23994df : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e9e7c73d020d0dbfc0600765206b68fee37fe6a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:32:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf7fbfb6017e472992dbcf326184cb81 : SUCCESS in 1h 48m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e050dab14ceb407db2b4a6f9891cffd0 : FAILURE in 1h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abd1cf9d40da01e9605f61e586d740393c4fac00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-08 21:28:35.000000000","message":"Patch Set 2: Code-Review-1\n\nWaiting for UCA/ussuri/arm64 on mirrors and for more arm64 nodes as we are getting long queues.","accounts_in_message":[],"_revision_number":2},{"id":"b99b62f574486c2a8404c60496420b0c42059d82","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 04:52:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e847051c799406cc9700eae89239801a7198530c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 05:43:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bd5481acafb94de8be35ffd86618e236 : SUCCESS in 27m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3d61554aafcb4aacab35af19ca041d14 : FAILURE in 29m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0bbc62ceb1fd497497cc4a59cce62ffb : SUCCESS in 23m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/54bc2a7b13c54b78aaab731c49d291fc : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cc2d67b292044c5bf3ab00df9e44a8e : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a6f72a9800a4e3db54e979d9ba54235 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8dc382845af4be9b98bd8542470cd53 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5a44f2b158a4d6aaa72e6993a7f6afc : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014f48b9713d4fdea669baa5afa76d28 : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb473e3f2e5349fcb8ab166b81781606 : SUCCESS in 33m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2db9e9869aeb412e9e1f91541a279850 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f95b991a0511f05f348723aa9ddd3e354a9c706d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 06:42:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f65e59f728644e8851eeb6aa1683251 : SUCCESS in 1h 49m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2d10e5b49aa466e8a8c18737f27eb67 : SUCCESS in 1h 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"064f47d58520f0cbc01d1e481f6e0177cae31aee","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 11:25:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c11a138fd5d8ede4c3ef8d5e0ac0db9801f275a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 12:38:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2370ec114b5275f7106219ee859531105f0a5d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 14:43:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6fc345f47d7a4d12af6a5c9e51e315f6 : SUCCESS in 31m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f44bdcba843b4df1b814541dbc7e0679 : SUCCESS in 46m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4304b80b23c24b0b833971c9abcc7c46 : SUCCESS in 1h 44m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3340a65006124500be6551d16d31ca46 : SUCCESS in 1h 42m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/44d26599971344bcad0a7de295abb412 : SUCCESS in 1h 57m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd23ab9c963c477bb2b0f818aedbf2d3 : SUCCESS in 38m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/098a44d788324e6585781ef3eed7c11d : SUCCESS in 46m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8b0abac2a9cb414b9878d0a5bca378ac : SUCCESS in 1h 25m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3b15c9ee88a43688dfeeb8e52839e5b : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221f2bfcf2834e5e8257bc20656a4587 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c86619238544b16a7889d5735d53b84 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e43a45eefc84c7f897b8979b7606d7d : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15923ffdf8d4cc2bb0bb86057f9dce6 : SUCCESS in 11m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a45e43eb9e204cb687a796c18394c43e : SUCCESS in 39m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84fa32bd788249aebbad96bfd79189ee : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3253d478c1c34392b79434f4cf69a4f3 : SUCCESS in 1h 28m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9a6ec9602f6441da0249c52033e0896 : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/289ffc21f78d412396d0551567457011 : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":4},{"id":"8373b6d01b66c71a88c4007032ab9ac53437f4e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 17:44:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4fa50f17095c4b0b86d0e355f075a314 : TIMED_OUT in 3h 03m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8804bac30fa8471caa7e4619b47cb806 : TIMED_OUT in 3h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"375557c30c4122aeb46404bc388178da38d0ad26","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 09:25:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c65193626656652c4717181fe7457d8b31325d1f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 09:30:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc3e88dbb768e09380f156200d6fa3b9b483ba0e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 09:41:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0fa3084914fbf36c05453dfadfac07f31b88173b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 09:43:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4cd48ba493af96ca4b62c0fffc4670edfbd6cc55","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 09:46:12.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0a2c7f95f6e899de70f3d40a19b6b4587b882e6d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 09:51:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"67e0c574c23ecce54703abc7055d94f220f9828a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 10:45:29.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"afd93b0c8db8a84920616cf6160bdcb8dcd7a365","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:52:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/35a6221e7c684753b416217526310970 : SUCCESS in 49m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb2bb50096664d3d9c1fb5647a5095e4 : FAILURE in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9b140c62a4e7741ba303f06c6963952fd5906c11","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 12:30:46.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"b21cbc2bba599dbe1970d789cd32bab4ebd6633b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 12:47:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a824129d3c4cafea4bc5b60ed820d8732d5f0de0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:43:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80aea6b03d544aa8bfc01abeb5960096 : SUCCESS in 55m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b84426ca75f4114b77b9d50f6d712f1 : SUCCESS in 50m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"21e55830d78d0f4919492ac214afd62f564e2e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:48:09.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6384dffe5ecd42678daaa7e2794ca784 : SUCCESS in 26m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/46c671910f3644648ae303cdabd430b7 : SUCCESS in 32m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/12452afa7c2c4367b95ad360cf38cbda : SUCCESS in 1h 22m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7531c5ac0c084131b777cd702573f78f : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1deb22a7ccff4ee7863ceb031890f6fd : SUCCESS in 1h 26m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd093b8a8e454c1c974acf2d2c6aa6df : SUCCESS in 29m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/de0fa4ecef0f4be9812cb2e3a0c11dea : SUCCESS in 26m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b67d2ec31fee4fd1988be061f538fa5c : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bae852e5c8384a63abeca6aa96954d26 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbcea2eb6f14efca8f4b0adfa87d51f : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d9e86bf0f0408994302b10bf90ae26 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f0fc482f8a74be485aacc340ae04f3f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412808d90bc34280b29f3e42a5ba7b04 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d178dbf434475ca2dd35ee92c4a09c : SUCCESS in 8m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3fc5fbb62c2941ed870644d6552fa498 : FAILURE in 50m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/86919863de744afca5ee7a311c18e624 : FAILURE in 49m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/318006094693451baaca51e2b803ee16 : SUCCESS in 1h 09m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d3070fa338a4f13aef7a64d6c872901 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/754d8121066147558ea66175f7013951 : SUCCESS in 1h 22m 03s","accounts_in_message":[],"_revision_number":9},{"id":"a0d7dd3487316896477effd6df6ed85d45bd53c6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 15:48:31.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"01c81adbfb350f21eb3be77fae5a21d4275dd0a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:26:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ebf60065b3e4879b2de66dde319a019 : SUCCESS in 57m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/66473834ce3b457ca4a91dd44f7efe0f : SUCCESS in 54m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b728a693fa330fa9ce6e50cbdbb0222541c453be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:30:54.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dd8ec068656f4664930449cb31890278 : SUCCESS in 30m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/847a8890b83a4760ad4a68c5eec99812 : SUCCESS in 28m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80854e4c38dc40c9802e9b6e77c7093e : SUCCESS in 1h 18m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/52b48b52805944ce96159debc5d2d605 : SUCCESS in 1h 28m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8c8609923f164a65895ee7a3deec3407 : SUCCESS in 1h 30m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c3933baa810499fadf9d3f4c39dcb6a : SUCCESS in 30m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/262f06a5ec5744ae8692e4ef38bf54d7 : SUCCESS in 19m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6276925b3e14dc2bdd719c77f7ed581 : SUCCESS in 1h 04m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/325a3bc56b1142d9ba45ea854171821c : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bfb4b14a484b9ab656a076b22b6fe2 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d43874cb1ac6425da8b1b1e87d0fe80d : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a66def01c6204375b161cdaeb7901d3a : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5efadc30302d49ecabd642cf985a701f : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc8d6f38bed4273a7b3cba6d56e1f55 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/704e9f20f6674f2589c7fae2cc35aff6 : FAILURE in 46m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e480666f8b7344ffbaf94ce349788446 : FAILURE in 35m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d31992c5f444d19b3802617c4c1c8e1 : SUCCESS in 1h 30m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1be33e44183c4948a7e10701579c8da2 : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/213d8d952cf34e87a601690943b242db : SUCCESS in 1h 51m 15s","accounts_in_message":[],"_revision_number":9},{"id":"7158f526234a6468f518350ced87b3fd786106af","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:11:44.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"31f37a10a2808b43bd1a4e9dc85e03a6d5738954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:19:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/49ce35cb72d84c21aef995c75a962f38 : SUCCESS in 50m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1838a23fd066468fa3121bcafdb8f1e6 : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3979a45c007b713819b6e2cfa222e8964364bb17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:01:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6f0a6e741da64f50a24a5ac3622838ee : SUCCESS in 50m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/72472a0af01348d394eb9fc3ee683f47 : SUCCESS in 54m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f7e936017a5f4b3095e67cdbed5654d4 : SUCCESS in 1h 22m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/320ce518c0f54fe694d2469e47ebd6b3 : SUCCESS in 1h 19m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/66ba4247d2aa4feda9021daada7e060d : SUCCESS in 1h 25m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/002a67fa0e1b4461b23d89f74df15cc3 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/184c455c6fe1430dad376b7b22baa352 : SUCCESS in 33m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3017fb93c1374d1a8910ba131119ec9b : SUCCESS in 1h 07m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c8a02e3faa24df9ab31122440592508 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43be1a17e1b44b13a8cc151e23b1d2b9 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0e6ed5ef60d45f280ee87a1f16096f7 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33f47a2f59a741e4810b72f74233b11a : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a92083231e145a7be8a0e3722c7472f : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1ae101384842a0b588ed75c4a0ff46 : SUCCESS in 6m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/626fe056484d4d9392951b6442324f02 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6ad6c98415241bd8c682462b32db6b8 : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf1671c19d724319973b333695a76599 : SUCCESS in 1h 06m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3480a4090044a60b42b01ca1a681ac6 : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/454cb7c8542d4b2eb59fc4112a2b6543 : SUCCESS in 1h 21m 01s","accounts_in_message":[],"_revision_number":10},{"id":"d47480e6569f44066776d9e57fc4ecd475cf9142","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-09 06:32:21.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"748a1c414d9ba9304c019f72550472a78b73bd06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 12:59:21.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"0fedd3378fb0e833e8e83d2c9f129918091b366f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:59:38.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"60b667226f02304fd334514db7e7d4540333c2ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:50:32.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d08041973adb49749c38f778bbd74faf : SUCCESS in 32m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/48f8be1580ed41cfbc81158a57c8fbd0 : SUCCESS in 1h 44m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e6a86680ef5c43c0aae1e6a492ad9558 : SUCCESS in 1h 44m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b20dece1b704486389e88575144b3f75 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20039fd148cb425f8e0bb613511df385 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7ad6936fcb414b9131797d42a712fa : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd24ba7f580e45899355da338a5a6ce4 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4835def4e6b9455a8b48eedbd33fa7e7 : SUCCESS in 6m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77d06d174b494b44b6c8bf9ac1456c2c : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4daac1d7caf549be937096b31ca389d5 : SUCCESS in 1h 02m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06782af142b243fda082527324634d41 : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":10},{"id":"4b31e636b5ac717ab788101cf6d443ede42ced6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:50:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"ec34cf62130123bb0b957620c9c8a9bb3cacccfd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:57:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52650fb01a874eefbbd902ae5eb16fd8 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"584da0675127d6b692a84bacfaa789b26a332ccb","revisions":{"fe9cb67b8a583af4101ce8bfc0e186f4a7af72cd":{"kind":"REWORK","_number":1,"created":"2020-01-05 21:18:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/1"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"subject":"Add CI run for Ubuntu on AArch64","message":"Add CI run for Ubuntu on AArch64\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9cb67b8a583af4101ce8bfc0e186f4a7af72cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9cb67b8a583af4101ce8bfc0e186f4a7af72cd"}]},"branch":"refs/heads/master"},"c5168eb4fedbfa2ce410664a0e57562919e2fef7":{"kind":"REWORK","_number":2,"created":"2020-01-06 13:40:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/2"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-06 13:40:25.000000000","tz":60},"subject":"Add CI run for Ubuntu on AArch64","message":"Add CI run for Ubuntu on AArch64\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5168eb4fedbfa2ce410664a0e57562919e2fef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5168eb4fedbfa2ce410664a0e57562919e2fef7"}]},"branch":"refs/heads/master"},"6eb7ee47ae0965892f022c75f2901f8d8b852e02":{"kind":"REWORK","_number":3,"created":"2020-01-10 11:25:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/3"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 11:25:00.000000000","tz":60},"subject":"Add CI run for Ubuntu on AArch64","message":"Add CI run for Ubuntu on AArch64\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eb7ee47ae0965892f022c75f2901f8d8b852e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eb7ee47ae0965892f022c75f2901f8d8b852e02"}]},"branch":"refs/heads/master"},"6a3fe3306734d0f6b57fa723777f6e07148957d4":{"kind":"REWORK","_number":4,"created":"2020-01-10 12:38:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/4"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 12:37:58.000000000","tz":60},"subject":"Add CI run for Ubuntu on AArch64","message":"Add CI run for Ubuntu on AArch64\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3fe3306734d0f6b57fa723777f6e07148957d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a3fe3306734d0f6b57fa723777f6e07148957d4"}]},"branch":"refs/heads/master"},"abeacd83f8e7fcc9b24582896f0b9785df42f380":{"kind":"REWORK","_number":5,"created":"2020-04-08 09:25:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/5"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:25:04.000000000","tz":120},"subject":"Add CI run for Ubuntu on AArch64","message":"Add CI run for Ubuntu on AArch64\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abeacd83f8e7fcc9b24582896f0b9785df42f380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abeacd83f8e7fcc9b24582896f0b9785df42f380"}]},"branch":"refs/heads/master"},"5c284f34eb5c334f739d1189bef7bd0dfb2bbc6c":{"kind":"REWORK","_number":6,"created":"2020-04-08 09:41:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/6"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:41:08.000000000","tz":120},"subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","message":"CI: add Centos 8 and Ubuntu jobs for aarch64\n\nalso simplify Debian job definition\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c284f34eb5c334f739d1189bef7bd0dfb2bbc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c284f34eb5c334f739d1189bef7bd0dfb2bbc6c"}]},"branch":"refs/heads/master"},"1e8f9280bd20885daf83de9d4f86a12bd659f073":{"kind":"REWORK","_number":7,"created":"2020-04-08 09:43:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/7"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:43:47.000000000","tz":120},"subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","message":"CI: add Centos 8 and Ubuntu jobs for aarch64\n\nalso simplify Debian job definition\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e8f9280bd20885daf83de9d4f86a12bd659f073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e8f9280bd20885daf83de9d4f86a12bd659f073"}]},"branch":"refs/heads/master"},"e993fca64c20639f1b894ac24d0c30923ea0836d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-04-08 10:45:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/8"}}},"commit":{"parents":[{"commit":"8a020dade96eae0781abf901557bc2f69cf01ba9","subject":"base/debian: pin erlang 22 for rabbitmq","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a020dade96eae0781abf901557bc2f69cf01ba9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 10:45:29.000000000","tz":0},"subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","message":"CI: add Centos 8 and Ubuntu jobs for aarch64\n\nalso simplify Debian job definition\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e993fca64c20639f1b894ac24d0c30923ea0836d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e993fca64c20639f1b894ac24d0c30923ea0836d"}]},"branch":"refs/heads/master"},"1eae0f41f3976dced35ba2fab49d53acd533d799":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-04-08 12:30:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/9"}}},"commit":{"parents":[{"commit":"8a020dade96eae0781abf901557bc2f69cf01ba9","subject":"base/debian: pin erlang 22 for rabbitmq","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a020dade96eae0781abf901557bc2f69cf01ba9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 12:30:40.000000000","tz":0},"subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","message":"CI: add Centos 8 and Ubuntu jobs for aarch64\n\nalso simplify Debian job definition\n\nDepends-on: https://review.opendev.org/718413\nDepends-on: https://review.opendev.org/718363\nDepends-on: https://review.opendev.org/718236\nDepends-on: https://review.opendev.org/718238\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eae0f41f3976dced35ba2fab49d53acd533d799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eae0f41f3976dced35ba2fab49d53acd533d799"}]},"branch":"refs/heads/master"},"584da0675127d6b692a84bacfaa789b26a332ccb":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-04-08 21:11:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/21/701121/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/21/701121/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/21/701121/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/21/701121/10"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-05 21:17:31.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 21:11:44.000000000","tz":0},"subject":"CI: add Centos 8 and Ubuntu jobs for aarch64","message":"CI: add Centos 8 and Ubuntu jobs for aarch64\n\nalso simplify Debian job definition\n\nDepends-on: https://review.opendev.org/718413\nDepends-on: https://review.opendev.org/718363\nDepends-on: https://review.opendev.org/718236\nDepends-on: https://review.opendev.org/718238\n\nChange-Id: I2c75dbc2ea063dd3507eaaf93a484c50acae5fb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584da0675127d6b692a84bacfaa789b26a332ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/584da0675127d6b692a84bacfaa789b26a332ccb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
