)]}'
{"id":"openstack%2Fkolla~701349","triplet_id":"openstack%2Fkolla~master~I1bd5a40eeef9612d995c81426fe510e89f438725","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I1bd5a40eeef9612d995c81426fe510e89f438725","subject":"horizon: fix build in stable branches","status":"MERGED","created":"2020-01-07 10:48:11.000000000","updated":"2020-01-07 19:37:17.000000000","submitted":"2020-01-07 19:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701349-1578425436909-1b68ee4e","meta_rev_id":"87f75b3d8956caa80c15272a616f85b8b3376b7c","_number":701349,"virtual_id_number":701349,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 19:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-07 17:07:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-07 16:54:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-07 16:21:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-01-07 16:54:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-07 16:21:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-07 16:21:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 16:19:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-07 16:54:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-07 17:07:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-07 19:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6b7ae6408386fc95eec5576874432df3a485dd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-07 10:48:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"02b5ca2b3256bd7e4cf37dacb09035d057b01b95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 10:50:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a70c6436b782de8e2bcf78bcd0f0c8ae26c5759a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 10:50:57.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"7d3f71c1c087e00141f789d46a3641e9ff1e9130","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 11:02:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9a7d4b40968c11777fb81315b9b50adbe3f2f998","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-07 11:44:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6745ec0b3fbf49d9bd61a7d10378221f : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9c0f79650fa81d283d306f93b1d599574531e133","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-07 11:55:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b331b5273a46dccdc71274050e6bab1a5ed11089","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-07 12:08:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5d6d6652e4f3c0f0c0ed403519178c83972cd609","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-07 12:37:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/d87351715a104b48b069c77b75656515 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":4},{"id":"13d72bef641637949a55397246a1fae38a1b8f0b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 13:36:27.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"a61244660818774879844173527e3dc8daaeb5f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 14:11:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e54b2ec823141aca3f8ae243c56dade : SUCCESS in 2h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dafa4816c75dadcffce94247b9960bd04646876b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 14:12:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3ea013e434ab43739e67754faf62404a : SUCCESS in 39m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2280237e26b64e07b10d0ab467fb8f27 : SUCCESS in 41m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7831e2a6f0da4f0e96d54f1af54d25fd : SUCCESS in 1h 31m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e47ba940cf7b48db9a8e409aa126a4c0 : SUCCESS in 1h 13m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9fdc5ec3b2214675a40c748c18a04e37 : SUCCESS in 1h 33m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f784026d9add4ef5b8e0d8929a2a6240 : SUCCESS in 49m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17f901ef822140d999035f37d387c33b : SUCCESS in 27m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c245d80bb369444487c1d84f734f9225 : SUCCESS in 19m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b4fae0d9edf240199f36245588559de9 : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5403036670344789b56be017b16baa32 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58d6f0578104b7fb93e0c6aee29eb55 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/272938890ad8424cb432cb362d5b6538 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46c07ffa20004355a8dfa85b2022e0e4 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65f8d90f9b841e28c64f842ebda2a13 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e080056078e84374ac076e84cfbcf45a : SUCCESS in 37m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f09f24d71d50422b8d6f48ea5729c3d6 : SUCCESS in 41m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7744cca421d446499eba01e2ddcbd1ce : SUCCESS in 1h 11m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7747223124fc4c6e8aaf8ec63ba8e2e0 : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62f4eaeb3ace4f33b193ea4a69ea54eb : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":4},{"id":"47b69584ebf93d2458232da3a69e1bd7ab8982c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 16:19:25.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f36ac2712d649245a112823d03b666a6178354fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 16:19:30.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"ffa90368b5a186e7195e984e58765cfd3a421047","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 16:21:15.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9c374291d188bf249eb7cbbdc27427c4234510cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-07 16:21:19.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ef62c865425608587edb894a061a75a146502b0e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 16:54:13.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d1755ab468fbc4e3d33032d42a54d68534179297","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-07 17:07:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e24e54d3be68411a9030d468f85e80fd : SUCCESS in 27m 58s","accounts_in_message":[],"_revision_number":6},{"id":"24e4463b4b398552812298bf21ec5c06c8009804","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:45:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74b44f43d4f14dabbfd13e8980feefb7 : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"36f455994d68de20474593004f1f9882e9deafcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:52:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7925cbc467ea40b5adab3302e7e835fa : SUCCESS in 43m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3e5ccc3b532d409c8e1b9ea74ed14501 : SUCCESS in 49m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b87b789e0bac43c2afc7631a8e73c322 : SUCCESS in 1h 21m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/91eecd7451d542f5bc0303219efbbb5d : SUCCESS in 1h 18m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/23c2b52881da4288ab26c95afa30285a : SUCCESS in 1h 30m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f2fed7d039af44489c4e6e406e9f7dbf : SUCCESS in 1h 00m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3863b55404e547db81b4e2971fa4c3aa : SUCCESS in 23m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/85fffa217b3e4a5c97663e72be33594c : SUCCESS in 29m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7db5d32ac1084264a9098ebbcd49353b : SUCCESS in 55m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4784b28556f84b1da6e610e536c0812c : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b37657419854312a487b9bf258fc708 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c46882f46a3414f99ea33116f4bf7cb : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c629cd341c7429aaf563497b2e6a678 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3ef4663634414293f90f46419393d7 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80b235e6b6fe4197ab73958de24741ca : SUCCESS in 31m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b57737be46354643998d43a32a81791d : SUCCESS in 31m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9f41c22cf6e43f0a4dbc02ad51bfa47 : SUCCESS in 53m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79a1f2ffd6e64097944d89b30a205c96 : SUCCESS in 56m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f70fc1b14b1e4927adadf79d86ba4d7b : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":6},{"id":"0c3e405fd5f3c209b6193a273578d33266176ee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:52:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"edd04274d4f667ceb5ab8763204660208281e17e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:30:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d164240f838d6292371fd58f31927444f7987043","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:30:36.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8b5da96b9ad24846ab1034c4a4e7c689 : SUCCESS in 42m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b2dede767a0a4477a40a4f39d473246c : SUCCESS in 1h 14m 03s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/af3a92103b4e4ee99ab9980f6217feb5 : SUCCESS in 1h 32m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c4c0dae62b14fbdbe57e22cd251c660 : SUCCESS in 44m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1424d0f664644e10b2291a73fd1873eb : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f0190bb4a640f788c7ad475b2d77e2 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6f33c55aab4485a32e6f6763639b73 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f133693fa244630afc8ffaa806b21e2 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327f7678eb3f43ca8579bfa4ae5ed750 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e79a7fb0a30149cea54478b5b522ec5a : SUCCESS in 29m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b80b3c16bca41d792388c1735a41131 : SUCCESS in 1h 05m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0e75c204f2b4b04b563887e686a409d : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":6},{"id":"be8b1fe4bb460cf1afcc70b3e1710eda14c06bf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:32:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20998d98dd8e45ffa7c8bee5faa29ccc : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":6},{"id":"87f75b3d8956caa80c15272a616f85b8b3376b7c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 19:37:17.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 54c98dc29d2764c099227b4d900e5b429e09368e","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8b7ec43536842dd500cc5ee1198a318070ac1046","revisions":{"dc701de7290344cb2adb4dbec6437e57740b4a2f":{"kind":"REWORK","_number":1,"created":"2020-01-07 10:48:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/701349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/1"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-07 10:48:11.000000000","tz":0},"subject":"horizon: fix build in CentOS7","message":"horizon: fix build in CentOS7\n\nSince [1] was committed horizon stopped building for train.\n\nThis patch removes horizon from upper-constraints, installs it, adds\nback and then processes plugins (which require horizon in u-c).\n\nForward-porting to master branch, since we use train on CentOS7.\n\n1. https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc701de7290344cb2adb4dbec6437e57740b4a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc701de7290344cb2adb4dbec6437e57740b4a2f"}]},"branch":"refs/heads/master"},"7f7082ba0bbf794e470b9e2a9cb9f7e0c3a580d7":{"kind":"REWORK","_number":2,"created":"2020-01-07 11:02:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/701349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/2"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 11:02:20.000000000","tz":0},"subject":"horizon: fix build in CentOS7","message":"horizon: fix build in CentOS7\n\nSince [1] was committed horizon stopped building for train.\n\nThis patch removes horizon from upper-constraints, installs it, adds\nback and then processes plugins (which require horizon in u-c).\n\nForward-porting to master branch, since we use train on CentOS7.\n\n1. https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f7082ba0bbf794e470b9e2a9cb9f7e0c3a580d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f7082ba0bbf794e470b9e2a9cb9f7e0c3a580d7"}]},"branch":"refs/heads/master"},"1177d2485b9d527b9ec43be1f3c67ee28ad86f2e":{"kind":"REWORK","_number":3,"created":"2020-01-07 11:55:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/701349/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/3"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-07 11:55:24.000000000","tz":0},"subject":"horizon: fix build in CentOS7","message":"horizon: fix build in CentOS7\n\nSince [1] was committed horizon stopped building for train.\n\nThis patch removes horizon from upper-constraints, installs it, adds\nback and then processes plugins (which require horizon in u-c).\n\nForward-porting to master branch, since we use train on CentOS7.\n\n1. https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1177d2485b9d527b9ec43be1f3c67ee28ad86f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1177d2485b9d527b9ec43be1f3c67ee28ad86f2e"}]},"branch":"refs/heads/master"},"8036dd4848b1e428ea940183219a3f63c4635560":{"kind":"REWORK","_number":4,"created":"2020-01-07 12:08:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/701349/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/4"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-07 12:08:00.000000000","tz":0},"subject":"horizon: fix build in CentOS7","message":"horizon: fix build in CentOS7\n\nSince [1] was committed horizon stopped building for train.\n\nThis patch removes horizon from upper-constraints, installs it, adds\nback and then processes plugins (which require horizon in u-c).\n\nForward-porting to master branch, since we use train on CentOS7.\n\n1. https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8036dd4848b1e428ea940183219a3f63c4635560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8036dd4848b1e428ea940183219a3f63c4635560"}]},"branch":"refs/heads/master"},"20612d3e788e2882b7f3a842cb39335848bfdb9c":{"kind":"REWORK","_number":5,"created":"2020-01-07 16:19:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/701349/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/5"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-07 16:19:20.000000000","tz":0},"subject":"horizon: fix build in stable branches","message":"horizon: fix build in stable branches\n\nWhen [1] was committed, horizon stopped building for Train.\n\nThis patch removes horizon from upper-constraints.\n\nKolla master is affected directly because we build Train\nfor CentOS 7 atm.\n\n[1] https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20612d3e788e2882b7f3a842cb39335848bfdb9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20612d3e788e2882b7f3a842cb39335848bfdb9c"}]},"branch":"refs/heads/master"},"8b7ec43536842dd500cc5ee1198a318070ac1046":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-01-07 16:21:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/49/701349/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/701349/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/701349/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/701349/6"}}},"commit":{"parents":[{"commit":"7c116653e311d2a9ef769bb9227acd0ce9d3ad7a","subject":"Merge \"Add \u0027--quiet\u0027 switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c116653e311d2a9ef769bb9227acd0ce9d3ad7a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-07 16:21:12.000000000","tz":0},"subject":"horizon: fix build in stable branches","message":"horizon: fix build in stable branches\n\nWhen [1] was committed, horizon stopped building for Train.\nAnalogous patch is proposed to other stable branches.\n\nThis patch removes horizon from upper-constraints.\n\nKolla master is affected directly because we build Train\nfor CentOS 7 atm.\n\n[1] https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b7ec43536842dd500cc5ee1198a318070ac1046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b7ec43536842dd500cc5ee1198a318070ac1046"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
