)]}'
{"id":"openstack%2Fkolla~701442","triplet_id":"openstack%2Fkolla~stable%2Frocky~I1bd5a40eeef9612d995c81426fe510e89f438725","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I1bd5a40eeef9612d995c81426fe510e89f438725","subject":"horizon: fix build in stable branches","status":"MERGED","created":"2020-01-07 19:37:40.000000000","updated":"2020-01-08 09:31:13.000000000","submitted":"2020-01-08 09:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701442-1578475873719-0e1c537b","meta_rev_id":"36cfac0799d571d2830f3e29f603f1134bb716a1","_number":701442,"virtual_id_number":701442,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 09:31:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-08 05:21:52.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-08 07:43:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2020-01-08 07:43:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-07 19:38:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 05:21:52.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-08 07:43:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-08 09:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd4ca27e7d4c39f64194980ed0cdadd0fdd014a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 19:37:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"bcfb9307ed2af05bee215f060ad85437fa487f82","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 19:38:37.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"0e7069b2fded13e1b2c5c9c50746b189cb41b9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:39:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/eab2ff8dc3b24caba0ad41f3976e7c9b : SUCCESS in 46m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b32a6b7189ac4e6eb9d9d8e8a17631fc : SUCCESS in 1h 01m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f109b03a45144b7f90cd7d899f8c5adb : SUCCESS in 1h 41m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/07d6f1960712423288c79849ab1c032e : SUCCESS in 1h 08m 54s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8fd994810e0844e9a87f90c58fa06db0 : SUCCESS in 57m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5a9cd468ac5848d6b6e2b3986f8a4168 : SUCCESS in 58m 55s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b44f4655e3724641b044ce423ea3c1cf : SUCCESS in 1h 48m 34s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3f98bb4a60e646aa9f33b06a21cf006e : FAILURE in 1h 00m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb23e1fe344444191c36765fc8573f2 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500d33a3887e4b05bef74584bf8fee86 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87400ca95a3b474a9e4add06b62cd0c6 : SUCCESS in 7m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/824e1094a2064dbb915704928a9a537d : SUCCESS in 6m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/76544cc8f45d43ab9e8c0ea8c8efa580 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4efe5ba434a4759868b7727776b2765 : SUCCESS in 1h 24m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a93355212f82466991343baabb535915 : SUCCESS in 33m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c122dc31b98641d284a0137e82fa08d2 : SUCCESS in 1h 03m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39872c1097544958839051a7efbc0206 : SUCCESS in 2h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b57e879ba1e28614d2d637bb4ebda40b3941ccb2","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-08 05:21:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f05a7f29762cdfb9fae2888a29e799519c803211","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-08 07:43:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20aa07e0aa01f14259f72d9b23d0eeece7804666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 07:43:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f90a054ae2984e6bba6078df619ba498cf6344b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 09:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36cfac0799d571d2830f3e29f603f1134bb716a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 09:31:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dd4f29dc2744481a81dcd7a5ccdee97b : SUCCESS in 49m 07s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c6b833de26f94dcf8786ec3831d50042 : SUCCESS in 45m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5e4ed3899224a48a16dd4c405210def : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d3cd753d2c41d98f08b8ea0693b958 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18fe652ab9f04f76a60a2361a8859760 : SUCCESS in 4m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f65de65549e64907a9c68c2e404b2509 : SUCCESS in 3m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5f061a03bca24271b85e93425ac73146 : SUCCESS in 5m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee88f198f5b04623a7eebeca5220136d : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bd33bdfc7a7f69d72f530e962acdfd00a539ea0","revisions":{"9bd33bdfc7a7f69d72f530e962acdfd00a539ea0":{"kind":"REWORK","_number":1,"created":"2020-01-07 19:37:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/701442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/701442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/701442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/701442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/701442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/701442/1"}}},"commit":{"parents":[{"commit":"595ed7987f4cd74bf940a58e6301ac902dd9b292","subject":"Bump versions for Openstack rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/595ed7987f4cd74bf940a58e6301ac902dd9b292"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 10:15:07.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-07 19:37:40.000000000","tz":0},"subject":"horizon: fix build in stable branches","message":"horizon: fix build in stable branches\n\nWhen [1] was committed, horizon stopped building for Train.\nAnalogous patch is proposed to other stable branches.\n\nThis patch removes horizon from upper-constraints.\n\nKolla master is affected directly because we build Train\nfor CentOS 7 atm.\n\n[1] https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65\n\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1bd5a40eeef9612d995c81426fe510e89f438725\n(cherry picked from commit 8b7ec43536842dd500cc5ee1198a318070ac1046)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bd33bdfc7a7f69d72f530e962acdfd00a539ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bd33bdfc7a7f69d72f530e962acdfd00a539ea0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
