)]}'
{"id":"openstack%2Fkolla~701748","triplet_id":"openstack%2Fkolla~master~Ie9a0109bd4eef337358d714e4b5088b8e3de2f23","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Ie9a0109bd4eef337358d714e4b5088b8e3de2f23","subject":"CentOS 8: Add zuul build and publish jobs","status":"MERGED","created":"2020-01-09 15:00:24.000000000","updated":"2020-01-10 12:47:56.000000000","submitted":"2020-01-10 12:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701748-1578660285238-46a14a57","meta_rev_id":"b4176ed781caec107e4f783121c23e48244e4b15","_number":701748,"virtual_id_number":701748,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 12:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-09 17:28:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-10 04:59:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-10 10:55:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-10 10:55:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-10 10:31:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 17:28:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-10 04:59:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-10 10:55:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-10 12:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4ba8226d69acd24ecc9fa3b93e8fabd10f5cb0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 15:00:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5134e39e5b4109102696c27758dc8d36f2024c6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 15:26:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/acbec2abe9fa44d39b2371693fcdf3fb : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"90eb4872ab744155d50f6ae440460827c91faf26","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-09 15:33:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2ea4e96da681ce2e0e77418d129cb00d1d05400","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-09 16:55:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa3071963c679783170d221702f6553249c514fa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-09 17:28:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e1d53786631c40b19c7193b8f0dc4eb0 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"083f9146209e6725ac6588d80714248d3fd8ec2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 18:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/619b83dd6095462ebcb39351d25802b6 : SUCCESS in 32m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a49e8840077a43e49679441bc0c233b2 : SUCCESS in 41m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f4b524b299424607a707c5b6f310cc84 : SUCCESS in 7m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/273322b201da408082b2e98eeb5a4631 : SUCCESS in 7m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fb4faf6b108a4446962c64b4f4275fea : SUCCESS in 1h 19m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/20cbad20d60e41b0a6e5f5d197b098af : SUCCESS in 1h 29m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ff6193df2bbf4d049c4858ca1665d353 : SUCCESS in 1h 32m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/abbe8f01b14c4bde939366388b28fb08 : SUCCESS in 44m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a73060c5b10c4fcaa6603ed52dcceeef : SUCCESS in 25m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e41daa0835c54c729302f80a0f17292c : SUCCESS in 23m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/404fc29734744e0ea5ef5849f52875c0 : SUCCESS in 57m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/376d3c77944a421eb225e94c1c8cb630 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec5323faff843389c9584b2bdc69eed : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e125ac80b4374701a601d89cff05efde : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6037621283b1473ebf0501faf00d6378 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886872fc701640a5947f2722b90df800 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9098905e4e154f81a5343c807fbed296 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/303e0a37867542dfbf6827f1a8fa86de : SUCCESS in 26m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70b3e3135e9949179a7b4ff2c43cf967 : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/690748df0d3f462ba01e4b1273ac7ff9 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8addf77563da4ffe8e884de659736a38 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f2c09da73b348ec9e85b9f17cdea727 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3b34a5b48bb0f397eb6b3d3f8915e614ba95366a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 19:51:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce226fd0d36d43fab58752add0f37d9d : SUCCESS in 2h 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd9b8017261d662a5aa2a2dcbea956581b12d5e1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 04:59:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c2bcce0a38ab5b284c96aa1eb3e6a484981bd6e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 10:31:37.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"80c40a43d812c7a30b80649ac71e7f2d6049966e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 10:55:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca5e88672c17e3fe49e3dc36fb1ba13399e815d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 10:56:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bda56b0a07c4f60de0fcc100e9f9b2512522ff8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:44:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/20f5cd86c5b84913a3cf9f2880e016a0 : SUCCESS in 38m 15s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a6cde2ff8b9c43e289e2f503f1ab1961 : SUCCESS in 6m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1fb5d5c766944aeeab3bdd594a1768c3 : SUCCESS in 1h 27m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e567314fbd4946ad974fdb45408dfe44 : SUCCESS in 1h 43m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/804c05577e9d4ca09fd0579c5396c721 : SUCCESS in 46m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d8c73755d514ef1900db2a4a20529c6 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1d22a04545474abde8ad78398aa421 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39fcc757b75e4d2a8649a445a2261624 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5be72fea8fb74e7a8032be3c6b174046 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3652de85c01432c80ad7d97f1c084f5 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2824f5e87f394b888d8a9d8e7d82a04e : SUCCESS in 5m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/813450f6f91f41d6968fff82da46cb1f : SUCCESS in 29m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4448e86a0848443189012b8a840bf61f : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/858b74add77c48919ea45b407e977a11 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9736d0b7ac7bddb6ca5cc6429d30a509a8d15450","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:44:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3c8ca31fc36dc36e8340f14348649dc77d702439","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:47:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c829867e02e24ee4ba4a6939f7e311a8 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e712fba4566241e99f2c8e4d34b9fd4f : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b060623b907348e518bdf1afede0b641be56c415","revisions":{"c52754929e9d2755b26140bc7a09a100e98e1e07":{"kind":"REWORK","_number":1,"created":"2020-01-09 15:00:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/701748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/701748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/701748/1"}}},"commit":{"parents":[{"commit":"71489044da447b61b65cc5e22f0020bd57fbc43e","subject":"Merge \"Add OVN bits to neutron-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71489044da447b61b65cc5e22f0020bd57fbc43e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 13:27:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 14:33:53.000000000","tz":0},"subject":"CentOS 8: Add zuul build and publish jobs","message":"CentOS 8: Add zuul build and publish jobs\n\nAdds build and publishing jobs for CentOS 8 images:\n\n- kolla-build-centos8-binary\n- kolla-build-centos8-source\n- kolla-publish-centos8-binary\n- kolla-publish-centos8-source\n\nThe \u0027base\u0027 image has been added to the list of unbuildable images for\nCentOS 8 (now separate from the list for CentOS 7). This means that no\nimages will be built - the jobs should essentially be a \u0027noop\u0027.\n\nChange-Id: Ie9a0109bd4eef337358d714e4b5088b8e3de2f23\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c52754929e9d2755b26140bc7a09a100e98e1e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c52754929e9d2755b26140bc7a09a100e98e1e07"}]},"branch":"refs/heads/master"},"b060623b907348e518bdf1afede0b641be56c415":{"kind":"REWORK","_number":2,"created":"2020-01-09 16:55:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/701748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/701748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/701748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/701748/2"}}},"commit":{"parents":[{"commit":"71489044da447b61b65cc5e22f0020bd57fbc43e","subject":"Merge \"Add OVN bits to neutron-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71489044da447b61b65cc5e22f0020bd57fbc43e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 13:27:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 16:23:39.000000000","tz":0},"subject":"CentOS 8: Add zuul build and publish jobs","message":"CentOS 8: Add zuul build and publish jobs\n\nAdds build and publishing jobs for CentOS 8 images:\n\n- kolla-build-centos8-binary\n- kolla-build-centos8-source\n- kolla-publish-centos8-binary\n- kolla-publish-centos8-source\n\nThe \u0027base\u0027 image has been added to the list of unbuildable images for\nCentOS 8 (now separate from the list for CentOS 7). This means that no\nimages will be built - the jobs should essentially be a \u0027noop\u0027.\n\nChange-Id: Ie9a0109bd4eef337358d714e4b5088b8e3de2f23\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b060623b907348e518bdf1afede0b641be56c415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b060623b907348e518bdf1afede0b641be56c415"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
