)]}'
{"id":"openstack%2Fkolla~701895","triplet_id":"openstack%2Fkolla~stable%2Frocky~I944e8e42fef1d359d767cbc6e1c13371ed753f31","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I944e8e42fef1d359d767cbc6e1c13371ed753f31","subject":"neutron: fix build in stable branches","status":"MERGED","created":"2020-01-10 07:55:22.000000000","updated":"2020-01-10 11:26:41.000000000","submitted":"2020-01-10 11:26:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701895-1578655601234-1b68ee4e","meta_rev_id":"9327123dc15faf13735fe1b76117abededc29994","_number":701895,"virtual_id_number":701895,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 11:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-01-10 07:56:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2020-01-10 07:56:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 07:56:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-10 11:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c1ea049a9d1ac39a4d7d7d2a93341ca22cf8684","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 07:55:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"baaeb3b9a6c1ca6db7bd39b27a2ceb47696b3bfd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 07:56:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nand done","accounts_in_message":[],"_revision_number":1},{"id":"40aa79ba4cf8436ea9502b5efa65fe7032c3fa35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 09:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f94415c0b8f04c2185f1755d49070ffb : SUCCESS in 46m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/de8c039fb41648cba8109d9fdfac52f6 : SUCCESS in 56m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71b7056907f84649839d288b37a9adba : SUCCESS in 1h 28m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9bc79c33e68a409083ae2ee68b98fdec : SUCCESS in 1h 23m 38s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cd0c8951aa1248828aaa352e2faa48a7 : SUCCESS in 54m 14s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6e1b4ce362e248009ba3a7bd2b93fe7f : SUCCESS in 51m 52s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0d364390eed8499c8bb9d0cfc9233617 : SUCCESS in 1h 29m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a4f270ca40d241568a65133501678b98 : SUCCESS in 1h 22m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fb2b68aa5874ee6a2b2c23c84eb8bb5 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5930b454bbd4d3e9450ba5b4f6471a7 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ecee0bfaa094db2871f651cd9dcffd4 : SUCCESS in 6m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93430ab54eb54ce699f1540d7f20bd37 : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dd5acd88e20a4791938ce01dee25049f : SUCCESS in 7m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/625bfbfcb1074e129638650bf84713b9 : SUCCESS in 1h 06m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0826c34f7ffd416b984d055859e6cd1e : SUCCESS in 41m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b1ac5e26fe7447a9477d5e1dd7dcbce : SUCCESS in 1h 13m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65c0488a71cd4d11a4555284589639f7 : SUCCESS in 1h 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f04f2e877fa7227ff98fe4d0dc43a91974eaedb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 09:59:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dba91d58296a1808c50ff2e6fcb0daf8e72c945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:26:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3d1904b27741454ea512004e6f376526 : SUCCESS in 57m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24251aa34b7c42dd8166bf194436145d : SUCCESS in 47m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dc8743582cf4559b33495764b0cb538 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1dee9f554574166aa3950bb7a6939cd : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49035f45cd6f476a99b48e4c53e12a07 : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e2e198e8b014116a7b2c2195fe52066 : SUCCESS in 4m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/308050fed0ba48f7b0a9d42ef8ea5858 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6d3f36b7f284ff08bd0b21c5b50a805 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9327123dc15faf13735fe1b76117abededc29994","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:26:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de9423225dad4be967b81104962a4c1cb2393dde","revisions":{"de9423225dad4be967b81104962a4c1cb2393dde":{"kind":"REWORK","_number":1,"created":"2020-01-10 07:55:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/95/701895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/701895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/701895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/701895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/701895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/701895/1"}}},"commit":{"parents":[{"commit":"9bd33bdfc7a7f69d72f530e962acdfd00a539ea0","subject":"horizon: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bd33bdfc7a7f69d72f530e962acdfd00a539ea0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 05:16:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 07:55:22.000000000","tz":0},"subject":"neutron: fix build in stable branches","message":"neutron: fix build in stable branches\n\nWhen [1] was committed, neutron stopped building for Train.\nAnalogous patch is proposed to other stable branches.\n\nThis patch removes neutron from upper-constraints.\n\nKolla master is affected directly because we build Train\nfor CentOS 7 atm.\n\n1. https://review.opendev.org/#/c/697370/\n\nChange-Id: I944e8e42fef1d359d767cbc6e1c13371ed753f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de9423225dad4be967b81104962a4c1cb2393dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de9423225dad4be967b81104962a4c1cb2393dde"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
