)]}'
{"id":"openstack%2Fkolla~701911","triplet_id":"openstack%2Fkolla~master~I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e","subject":"CI: Fix symlinks for failed build logs","status":"MERGED","created":"2020-01-10 09:38:38.000000000","updated":"2020-01-15 09:53:18.000000000","submitted":"2020-01-13 00:07:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701911-1578874053794-bbef5f2b","meta_rev_id":"8b30010389493673d6cad0bed2b160bbe1608617","_number":701911,"virtual_id_number":701911,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-13 00:07:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-01-10 09:41:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-01-12 22:36:52.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-01-12 22:36:52.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 09:41:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-12 22:36:52.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-13 00:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a3a0cd159ba24b8317f6c8258aa7fdafe3be868","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 09:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"673a7b94fbe15b51f28596b218b0680e05292847","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 09:41:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf8adfe59f4bec7f37dd944a487bb5c72354d871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:08:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9feeb9723274343bab32b0234c342e0 : SUCCESS in 39m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4093b43ca05a4ab694ca813f7b2e07fe : SUCCESS in 37m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/98dfad8b83314802a581d1a8f8dc1c69 : SUCCESS in 1h 15m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/373edfb9fc004376a6092fb943c77224 : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1212a65a03ac476084039531e3a85cf9 : SUCCESS in 1h 28m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/165dd80004944b068eb8df3b22a65907 : SUCCESS in 26m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0913a7a101db4204932fcc5f6c47e547 : SUCCESS in 19m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/35bc44144668485da2fb662cac25c0ac : SUCCESS in 53m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef4253656124059b575894431f8212b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fddf8c8c19244880be793a248672a540 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5743e989737a47bfb4bdc0db33fe60b0 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdf7485750b2457ca249834cb82dc9db : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942edd203bb34c3383f38f546439d34f : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0418a895b0b94875a60fc3b8514b7439 : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba9769106d1f4f008efec0e46ead715b : SUCCESS in 28m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33de78d7b330461e90906d4f5872cd02 : SUCCESS in 52m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf24861c26294e64aae898a3457e1da3 : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a2537b7c83e49a3b2db13862a4eb039 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":1},{"id":"73c342a3a4e2dbfd74cce71843f316e87a4f6959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47f8ec19999a410e87e4ded2bbf41845 : SUCCESS in 2h 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e34810b8e5889a2543b86130f150692b8ca3afa6","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-12 22:36:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6fe767547f941b5bf8c21ff6e307f9e4eb946e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 22:37:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32eb5ac8e61733a89ca21d3b20a79d5d9d0eae76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 00:07:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc47d60433fc75931e47045eb803d45ad1291854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 00:07:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bad9c6b493f344d88c4de1c44db6565e : SUCCESS in 40m 08s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a0c691b331be4967a9123183a9fe5c33 : SUCCESS in 5m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f4f247e0229547ddbf9d6a84bfa5b181 : SUCCESS in 1h 18m 47s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0112b1d557904af5ad2c74893b3d5f5d : SUCCESS in 1h 29m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58310a5f01b4c69b58f08a08ca3acce : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b769e56698c418e9a1a415c152d2334 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62910ea5b3ba48929e095e2834b615ed : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64e40420016c4eee95aeb445758562c0 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94b521248ad41ddb36caa3846e2f92a : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/941fef7e4b064d35ba12a9ca080bef44 : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8f83866c9b0461eb864c41eb5ef1e38 : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d19f62759525472b945cb9a98efe651e : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c4a64f197fb29fdd6e28b507a6ec2902d3b94276","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 00:08:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e5478565b4446991d346fb2709962e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1},{"id":"129b59cc040c23ecb49d821fbd0b7f27d198a89f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-13 10:02:20.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 583df1510c576bb8b306e9dba4101c0ad3fe3bf0","accounts_in_message":[],"_revision_number":1},{"id":"5fff96cda9d3c4a20c28a36bf2a039eff95ce99b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-13 10:02:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f6ad36b5fe56ec2d80e703751ae641b02d98b99b","accounts_in_message":[],"_revision_number":1},{"id":"c3c8e9d0e3b73b5ec6f0032bc94d7a0bff0ade3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 09:49:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b30010389493673d6cad0bed2b160bbe1608617","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 09:53:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63f773f569c7601deab40b32bd62ccde513fa46f","revisions":{"63f773f569c7601deab40b32bd62ccde513fa46f":{"kind":"REWORK","_number":1,"created":"2020-01-10 09:38:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/701911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/701911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/701911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/701911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/701911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/701911/1"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-10 09:37:09.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-10 09:37:31.000000000","tz":0},"subject":"CI: Fix symlinks for failed build logs","message":"CI: Fix symlinks for failed build logs\n\nFollow up to I5b5f323471676317a2898875cbbf297082328fcc. Logs are no\nlonger gzipped, so we do not need to modify the 000_FAILED_* symlinks.\n\nChange-Id: I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63f773f569c7601deab40b32bd62ccde513fa46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63f773f569c7601deab40b32bd62ccde513fa46f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
