)]}'
{"id":"openstack%2Fkolla~701914","triplet_id":"openstack%2Fkolla~stable%2Frocky~I5b5f323471676317a2898875cbbf297082328fcc","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I5b5f323471676317a2898875cbbf297082328fcc","subject":"Stop gzipping logs in post","status":"MERGED","created":"2020-01-10 09:40:56.000000000","updated":"2020-01-10 13:32:14.000000000","submitted":"2020-01-10 13:32:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701914-1578663134002-c94174cc","meta_rev_id":"1d083097984040c6a1b59d08b9c3d2849320c612","_number":701914,"virtual_id_number":701914,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 13:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-10 12:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-10 12:23:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-10 12:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 12:31:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-10 13:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40855d3851ab1d58540c47c875dc2b8c8e9140f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 09:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2887fee68c891a606bf303680fa19ae0d3fb6a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a14a945300ed4bf1a8e4625dd2051aef : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a1bbcbe0d77242b6afc155c6160a137b : SUCCESS in 58m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3ceb3c18937d4a16b370b1d3f9bad3cf : SUCCESS in 2h 01m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4b56d64e604e46d5bb659f31c0fe26b9 : SUCCESS in 1h 15m 47s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4e4aba19492a4ba997406897be6be4a6 : SUCCESS in 52m 30s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fc30e90d38d3401296acad91032461b5 : SUCCESS in 1h 12m 21s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8bcc2cad43c04185bae15f070bff5992 : SUCCESS in 1h 22m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/36c829555f484d6293b2bb00f4ddb758 : SUCCESS in 1h 15m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f75d4dff9194d9d904ef162f2d2c027 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9166339f95471a92e3e30e5af2b0de : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e003739a11048d99ea5bfd64e80d184 : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0787ba6393324d76abb0c067ed0fb24f : SUCCESS in 5m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7e3e084cd32e457a8db3f6dc49154078 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76950b23d4c947f4a080bf9d305948a4 : SUCCESS in 1h 29m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8481be5e69d748de96563dada833d9ca : SUCCESS in 49m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f35829fc31547468fba4f96556d0a8d : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20231aa8176a4400b4f441e5c5b2b86f : SUCCESS in 1h 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38732df9e76ee7bff97ad8c4f8c76db67279410f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-10 12:23:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6b9a032b21c9fcdaa1ffaed1f3626e91393ca9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-10 12:31:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d86250696e23629a9bf508c9ea62ef9e33f56dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 12:31:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80d19a1ce5fa1d758ed94502f70f38737c3ed6ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:32:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ee07e289faf44a969bb75d58494eca3b : SUCCESS in 53m 54s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1974c4d6c37f4be4b4dbc774a59e07ed : SUCCESS in 45m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0369ae055df4a9fa8433336903afa4d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef00714206ec4b32808382ee50595a8a : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eaf1755054b4083b9f85a3354749a1b : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d99a215b51ba4f78855feb2c7945c275 : SUCCESS in 4m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7ccc6cc78d284d1aaafbb0c02d955e99 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ea6938b9bc3461f891d06084624553f : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1d083097984040c6a1b59d08b9c3d2849320c612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:32:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3709ded397aea5b03f43ea6812b2ba0afd74ed5e","revisions":{"3709ded397aea5b03f43ea6812b2ba0afd74ed5e":{"kind":"REWORK","_number":1,"created":"2020-01-10 09:40:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/701914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/701914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/701914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/701914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/701914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/701914/1"}}},"commit":{"parents":[{"commit":"9bd33bdfc7a7f69d72f530e962acdfd00a539ea0","subject":"horizon: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bd33bdfc7a7f69d72f530e962acdfd00a539ea0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-09 15:33:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-10 09:40:46.000000000","tz":0},"subject":"Stop gzipping logs in post","message":"Stop gzipping logs in post\n\nAs per [1] we should stop compressing the logs sent to swift in order\nto get them back readable via a browser.\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: I5b5f323471676317a2898875cbbf297082328fcc\n(cherry picked from commit f44b0b778b0d3a0bfa2f7d5cc126618c4c1e67f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3709ded397aea5b03f43ea6812b2ba0afd74ed5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3709ded397aea5b03f43ea6812b2ba0afd74ed5e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
