)]}'
{"id":"openstack%2Fkolla~701986","triplet_id":"openstack%2Fkolla~master~Id3b55ac470a977be7ee7decdef3b49e8d89a130f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Id3b55ac470a977be7ee7decdef3b49e8d89a130f","subject":"Use Debian mirror in CI jobs","status":"MERGED","created":"2020-01-10 15:42:44.000000000","updated":"2020-01-13 10:17:19.000000000","submitted":"2020-01-13 10:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701986-1578910536925-69f755c7","meta_rev_id":"b2f839078f888e0822d92a7d38412e5da66258e7","_number":701986,"virtual_id_number":701986,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-13 10:15:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-13 08:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-11 10:20:19.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-13 08:36:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-13 08:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-11 10:20:19.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-13 08:28:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-13 08:36:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-13 10:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a924b7888b6b04a1966e9c1348147555a23c9b3f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 15:42:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8058b5b56159b8e68fc2760272b5847a12e39a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 17:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/430d3575ada84da3898e2644d8bf272d : SUCCESS in 35m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6849c30d8bfb477692be8a18987ddd32 : SUCCESS in 45m 24s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/39901e53cf6e49cf89b9dd2934975243 : SUCCESS in 5m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a1c1f17fc1b84b84904ed0d1bb0a8253 : SUCCESS in 6m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7f27f118d02e4f36ab16fed9116f3528 : SUCCESS in 1h 39m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/79f2539648314efb84a2d1623c4cffa3 : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1d5728a763a34a62bbe3072025cd8886 : SUCCESS in 1h 51m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6a2ea67f3a024c098a4e77d8cfef71a5 : SUCCESS in 30m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1cf3ed56567f48049bed819b0ba782e7 : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7edcdb3a732d4041a2b745dcd569eeb3 : SUCCESS in 1h 01m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4e9bbd9586a4f329def551b11b808ea : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77808d03d82843b59a85152f6c954864 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f84fdda9cbe444b1bd2cf90feaa4e683 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/baeb37b2098d4f0f8f30408763053573 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6717dc9b5d30455c985e46b0c53b9ced : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7c37a06409544b8b2621c8519e90ef7 : SUCCESS in 27m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b64d093c0f74788840d83963f7fec9d : SUCCESS in 38m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fd208dba77e4557ba8f96b078c6fb88 : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7360888d27bc42c0aa67dccfa653960b : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0a9a4ddd639467b9c7753c2319b0a87 : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ca1912855f3c4851148396097a0f408801a041b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 20:51:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/686782731d5642e1aca39173d65c788b : TIMED_OUT in 3h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f86a978b9e2bc445eb2309378146218893fd948e","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-11 10:20:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"833348df3ad5782e5942c4948177c59d1dd87366","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-13 08:28:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ff8dd84a38545484a51a3ac24cfeaa31d931c19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 08:28:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fd12d7d4f7e6bd891c31a8fe267531d344d6b05","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-13 08:36:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9842cde8f62de24e09821200553f4590e03549d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:15:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1629f19d376a5473f6b357dfa04b4511b28964f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:15:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b7479591d3584300b8394b76d3f69bfc : SUCCESS in 43m 07s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/414b1da77aca423e93811e859f473a8c : SUCCESS in 6m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/85522d1614474e6f800d5f667ba8c343 : SUCCESS in 1h 29m 15s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e68600068ddb429a8d20ac5c6aaac11a : SUCCESS in 1h 44m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b74b23b161240c0b13fcfef7d535a0f : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4965d1bc3524f33ba59ef9cc8c76c95 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc5b86940c844275ac4800edd2560682 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72f64509bd204f6abbd5329c4d835230 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc37bf602399445891dbec4727ecc19c : SUCCESS in 8m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c96780de2f64ad4b76387aeb84436c1 : SUCCESS in 33m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d9cba0b1a8a41119c3bfe542c58c6a7 : SUCCESS in 53m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bd2e039233644d7be2b81da7960296a : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b2f839078f888e0822d92a7d38412e5da66258e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:17:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c7bb1f535244d1af298d639e4b876c : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cf3f2309a2a774f65d403b90d1950cab001863f","revisions":{"2cf3f2309a2a774f65d403b90d1950cab001863f":{"kind":"REWORK","_number":1,"created":"2020-01-10 15:42:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/86/701986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/701986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/701986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/701986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/701986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/701986/1"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:40:50.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:42:32.000000000","tz":60},"subject":"Use Debian mirror in CI jobs","message":"Use Debian mirror in CI jobs\n\nThere is Debian mirror on each CI worker so we should use it.\n\nChange-Id: Id3b55ac470a977be7ee7decdef3b49e8d89a130f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf3f2309a2a774f65d403b90d1950cab001863f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf3f2309a2a774f65d403b90d1950cab001863f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
