)]}'
{"id":"openstack%2Fkolla~701987","triplet_id":"openstack%2Fkolla~master~Icb745f628c1c50d9f20df022b08ddc46fccc270c","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Icb745f628c1c50d9f20df022b08ddc46fccc270c","subject":"CI: use Debian/Ubuntu mirrors in base image","status":"MERGED","created":"2020-01-10 15:47:58.000000000","updated":"2020-01-13 10:15:38.000000000","submitted":"2020-01-13 10:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701987-1578910538400-f3637170","meta_rev_id":"56bee332e3b1d64346ca07e7f9f5c5bf24b0f683","_number":701987,"virtual_id_number":701987,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-10 23:30:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-13 10:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-10 16:23:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-10 16:11:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-01-11 06:43:02.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-01-11 06:43:02.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 16:11:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-10 16:23:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-11 06:43:02.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-13 10:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b5ff7a9936beb88d877a9102bc0466f452af4bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 15:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"565c74103ef8683f91fae42aefeb4f018b20e5c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 15:54:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af27974f94c59fa9791a9c790a280c7c0f8102c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 15:54:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18529d7bf9d6547982312595db756fe6c8e31461","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 15:57:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a73f487c5f580a3b0b631cfd0ebcb80f5b96a623","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 16:11:30.000000000","message":"Patch Set 2: Code-Review+2\n\nnice, could help with stability (all depends on infra now)","accounts_in_message":[],"_revision_number":2},{"id":"c7fbc11ecf3bb2997116e93115caa68497a8d0c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-10 16:23:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/4f197d08688942f4a8a4ddfd5a589b0a : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d774160e40bc495ccb9b17be4b615cfcb5d4d50f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 16:44:21.000000000","message":"Patch Set 2:\n\n\u003e nice, could help with stability (all depends on infra now)\n\nwe install some packages from upstream, then replace sources.list, switch to infra mirrors, install rest","accounts_in_message":[],"_revision_number":2},{"id":"3e2328822ced97a93cc91ed70719ec597c19e4a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 18:33:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f3a0a92b327949149f51efb04ea556a2 : SUCCESS in 40m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/24587a2f14934114a01def8628f6949c : SUCCESS in 40m 12s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cbc440d8789d4799842d4048bfcbee15 : SUCCESS in 7m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/21e6c387d66748b589652a0eed55e4c5 : SUCCESS in 6m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc847d247f8242d8a8b021f9b2510d8e : SUCCESS in 1h 20m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6f28352ac3a94b068f044177c7640e4b : SUCCESS in 1h 22m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e60bb8a526ae431b8df39085c6895203 : SUCCESS in 1h 44m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/80dc131e5e524c858d6ed5c96bd8529b : SUCCESS in 1h 03m 54s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/22e3d89f31d541399c261155582334e5 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c0deee9d68d64ed09f9c69d10d433fef : SUCCESS in 2h 20m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0f74016d2f524031a6464845614dc361 : SUCCESS in 2h 29m 58s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0fd98654e6af492bb6bf9d67ca5d714d : SUCCESS in 1h 50m 48s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1c76edacbf7a427f98af0cec9a6f4d61 : FAILURE in 1h 10m 52s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/07a1e56bf2204f4b8e044a5b07212a8d : SUCCESS in 1h 23m 25s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ff827ee8175a4359b4a4f43deae5e15b : SUCCESS in 27m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1e37e9ce3ee545f0a1e64d113b8c4e98 : SUCCESS in 19m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba13684d8d144089b0d00e0b7d085725 : SUCCESS in 56m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c352edab566443daad4bf2a1a497c7a : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20efd58f00647b48f32a1c2d3ed2cb2 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcdeddf80e2b4c03b7aeb99119263074 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78547a342473491f98ae9d8d728e7d47 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b1137d59e0400093c1858d4cf2cd52 : SUCCESS in 6m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e275471eb14f405d8d1f9b0e180281d6 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe17b0ca0bcf4c128839702d46a6b833 : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7e8b16c25714e0ba007d84d7f68e009 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c1c75265823f478c8c527c076f275992 : SUCCESS in 55m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/572374fa5ea14197a4990af936223c4a : SUCCESS in 1h 02m 29s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/355e3d253c3344b4b29948912ba68109 : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/28c6e7b449af4b809397551875ea34fa : SUCCESS in 2h 16m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1a91c8bb272b4cde9e87e4f499a3c77e : SUCCESS in 2h 25m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/08976bfe60594889b980157382bdb6eb : SUCCESS in 1h 13m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cc1dc97ee1a247a5af07f2027b3daf57 : SUCCESS in 58m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2984439e8f6758344309b49cec99fea29198e90","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-10 18:47:10.000000000","message":"Patch Set 2:\n\n\u003e \u003e nice, could help with stability (all depends on infra now)\n \u003e \n \u003e we install some packages from upstream, then replace sources.list,\n \u003e switch to infra mirrors, install rest\n\nStill? Then the TODO should not really go away.","accounts_in_message":[],"_revision_number":2},{"id":"133d742df01c43a9f5901da016e4049f5c06a51c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 22:25:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e0c8e0f306db48f7ad7c0559b9fc4f7d : SUCCESS in 2h 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b7051f3cfa85db12107509a1c8eab300a81d73d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-10 23:30:47.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e nice, could help with stability (all depends on infra now)\n \u003e \u003e\n \u003e \u003e we install some packages from upstream, then replace\n \u003e sources.list,\n \u003e \u003e switch to infra mirrors, install rest\n \u003e \n \u003e Still? Then the TODO should not really go away.\n\nWe use repos from original container to install packages enabling us to use https repos. We also install \u0027lsb-release\u0027 and locale stuff.\n\nThen we put our sources.list in image, mangle it to use CI mirror and continue with work.\n\nTo do everything from CI mirror we would need to mangle repo urls twice.","accounts_in_message":[],"_revision_number":2},{"id":"e29dd33d2593e1343111b1f7ed1480083e3fd4e3","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-11 06:40:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7d0007ddf335b39991a62a9d43a1b515bd069991","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-11 06:43:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6bec3d2cd0efaf2170d6f0b214fe761f95fb193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 08:28:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4aba129d221fa9657b2d89389b450934635f45da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:15:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"56bee332e3b1d64346ca07e7f9f5c5bf24b0f683","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:15:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/af7632a5699d498c8f0e8da8c6330fdf : SUCCESS in 37m 56s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b7007cfbbb44f398959add8795f4caf : SUCCESS in 5m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cafce1c7b372431ca11848521f9c63c8 : SUCCESS in 1h 15m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/165414eaaff844a29e472edd633f2f90 : SUCCESS in 1h 37m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/25a6020de2b14046b6ea05e45dd3166b : SUCCESS in 53m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18bfd9dad5d643c5af7e4088e9686ab5 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2ed5eba9dd465bbd7d334579ae7a44 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc924bdc733c4fb8b83286a53825dc5f : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fe00195cfb5478fa567d9e6bbbfc1c4 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5922d6aef6dd404c82f140f272538e43 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3b3aa7398ca47e7a2373f9c71d3d81c : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc413d8a370c4b0391838e31e4da6805 : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/575247a26f0f4d99a11c7f66d3a62b84 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad03d7eaa25c31b201ec1c7dc2596f67bbe4fdaf","revisions":{"0f78f0e921b9f04260c39946ee688f22a50f21f0":{"kind":"REWORK","_number":1,"created":"2020-01-10 15:47:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/701987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/701987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/701987/1"}}},"commit":{"parents":[{"commit":"2cf3f2309a2a774f65d403b90d1950cab001863f","subject":"Use Debian mirror in CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf3f2309a2a774f65d403b90d1950cab001863f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:45:13.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:47:37.000000000","tz":60},"subject":"CI: use Debian/Ubuntu mirrors in base image","message":"CI: use Debian/Ubuntu mirrors in base image\n\nSo far we built \u0027base\u0027 image using upstream mirrors and then replaced\nthem with OpenStack infra ones. Let do that right after we copy own\nAPT sources.list file.\n\nChange-Id: Icb745f628c1c50d9f20df022b08ddc46fccc270c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f78f0e921b9f04260c39946ee688f22a50f21f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f78f0e921b9f04260c39946ee688f22a50f21f0"}]},"branch":"refs/heads/master"},"ad03d7eaa25c31b201ec1c7dc2596f67bbe4fdaf":{"kind":"REWORK","_number":2,"created":"2020-01-10 15:57:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/701987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/701987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/701987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/701987/2"}}},"commit":{"parents":[{"commit":"2cf3f2309a2a774f65d403b90d1950cab001863f","subject":"Use Debian mirror in CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf3f2309a2a774f65d403b90d1950cab001863f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:45:13.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:57:15.000000000","tz":0},"subject":"CI: use Debian/Ubuntu mirrors in base image","message":"CI: use Debian/Ubuntu mirrors in base image\n\nSo far we built \u0027base\u0027 image using upstream mirrors and then replaced\nthem with OpenStack infra ones. Let do that right after we copy own\nAPT sources.list file.\n\nChange-Id: Icb745f628c1c50d9f20df022b08ddc46fccc270c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad03d7eaa25c31b201ec1c7dc2596f67bbe4fdaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad03d7eaa25c31b201ec1c7dc2596f67bbe4fdaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
