)]}'
{"id":"openstack%2Fkolla~702180","triplet_id":"openstack%2Fkolla~stable%2Frocky~I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e","subject":"CI: Fix symlinks for failed build logs","status":"MERGED","created":"2020-01-13 10:04:22.000000000","updated":"2020-01-16 12:23:05.000000000","submitted":"2020-01-16 12:23:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702180-1579177385424-9981773a","meta_rev_id":"21a48281265fb2c6292f0ba7a25a0fa1256b0b55","_number":702180,"virtual_id_number":702180,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 12:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-15 09:46:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-13 13:12:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-15 16:58:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-15 09:45:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 09:46:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-15 16:58:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-16 12:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f895b211e1a42a6dae0896b1b16d364bc88398be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-13 10:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"612e58ec5fb6d873252c9817e4c90c64b70dc69c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 12:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5603dc33f76146d28ea9f2aa6a279223 : SUCCESS in 39m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a6c64e46f9a4477b104316b30d08610 : SUCCESS in 1h 08m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/882599b5ede8493c8c2e594b71e7f50f : SUCCESS in 1h 28m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59912782e6f74b1d88704fcc4ea56747 : SUCCESS in 1h 27m 41s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5a6266c2e38848b6b120d7892c163572 : SUCCESS in 53m 22s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7e69cce335ae4eb28c03c7702d9e9fd2 : SUCCESS in 59m 08s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/056d00c7498f440987244348497fba13 : SUCCESS in 1h 29m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/25b0a29095d24370ac4604169bf57ada : SUCCESS in 1h 24m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e5a1d482db84822a8a738b536606bc7 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff9ddfae74b4de0abbc29ae6b47c69a : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba6b2d82809e42dda4e9b58b3923ceaa : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ab548c2c0df42e1909564eb29814849 : SUCCESS in 4m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7de7aceb0bdb4689a4b352801daf3789 : RETRY_LIMIT in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12f881aa0d7648b8ba6c0968e2f78845 : SUCCESS in 1h 28m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29156243c0e2447ca77987de2f0f18c8 : SUCCESS in 29m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5dc5f9285434ae9860f2c2178d637b8 : SUCCESS in 1h 00m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f818f21e4454e4d8c6327ca5635e44b : TIMED_OUT in 2h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a234ac7e5ea1a974469d4b24857c4a1e3c1bf016","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-13 13:12:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9ef22819c4cda3ea0d217dd9aae4f413686dab9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-15 09:45:49.000000000","message":"Patch Set 1: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9e380c0a43746037ea9ef52fc1f773428d0de97","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-15 09:46:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd511a943fd6248e1352842375e330f6401338e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:18:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/16a9bfefb0a04974af4d75bcdf2a7b53 : SUCCESS in 39m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b03a7bf403e14c4e9f87e7d887a8d7bc : SUCCESS in 46m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ec521e52448841d3b7d3148f565c0f99 : FAILURE in 8m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8166640d0ab143338169ed27ed9086f3 : FAILURE in 6m 53s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/25db42a128474b7ea35b32150d3f172d : SUCCESS in 55m 10s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f1a68fa4638f483ab8159e508931855f : SUCCESS in 1h 00m 37s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2ad8d4ae13b940a09e1f073084fa365b : FAILURE in 7m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/175ce259e6c346d6be74f47784afc770 : FAILURE in 7m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59d795412ec41aaa52cb8e7e665f49b : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74aa99af7bbd49bba2943e90c4b4444d : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e12343ca72f1422f87fe1053f2727d31 : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a47f6ec07cbe4e2b95f43fac808d7086 : SUCCESS in 6m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a9c8684338c5452e9aefeac8ea3b018f : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ce4cc4193854e87a27571fc579fc39d : SUCCESS in 1h 27m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78947d643dbb431ea63d7421f3d47b5f : SUCCESS in 36m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa5b8af277a746f49d8e992c16460af1 : FAILURE in 6m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd07227cfa0c4595aa4db7a31b718d48 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0b4e58c729c0fa403b1d402e7f362d4948eaf5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 12:29:48.000000000","message":"Patch Set 1:\n\n\"msg\": \"failed to fetch key at https://yum.dockerproject.org/gpg , error was: HTTP Error 403: Forbidden\"\nargh v2","accounts_in_message":[],"_revision_number":1},{"id":"ba83237be1cbe67131b19f947769c61ea43e470d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 16:58:05.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ee5c4e84593155d2c335ce7bf5bddf0b2584080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 19:21:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1a5f8afe2be6442ca4b5981572a238b8 : SUCCESS in 43m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1c6b7ff929274379b56bca7c1c398ae7 : SUCCESS in 49m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f1323474afe6481595a4338754f88f17 : FAILURE in 55m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bd55624ba6bd42d693d9bf8fa6313e77 : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/35af85c633d442f09e7c29bb311219e2 : SUCCESS in 41m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/44ccb1b5ee694b0ea5c97dd92a1c0b67 : SUCCESS in 57m 17s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/565695b8c73b48e9be96b417e0dbc61c : SUCCESS in 1h 31m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fd63377145fd439fb34f4bf55100e698 : SUCCESS in 1h 10m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/415608fb7a354c80a3eb1fec6a6bf722 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b65c02689b43cebfcbb02187be1e84 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69873e0d083746d78fe68498024bc33e : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b46bfe60adc544438c753c64e8909c11 : SUCCESS in 4m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/96a5b10b607646cf97e03e69583021b9 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9877b271fc64646a518ae8f74f346ac : SUCCESS in 1h 27m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b7421c196f74761b20d99720b48fdf7 : SUCCESS in 46m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a60a637595145ae8ee036cc825ea43d : SUCCESS in 1h 15m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/128cee4b89584e368096c9650851280b : SUCCESS in 1h 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e77686dd1bc11c405cfc39d820c0781a6efdd886","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-16 09:28:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb99a9e14382e337d49b693c279a8e3597ccddca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 11:26:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f60371f648e74076ad9e5b97f95ac1a1 : SUCCESS in 46m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d005fafe66a447c2abff1fc853413447 : SUCCESS in 1h 02m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a2cff3173cee48a997d394b16c48c804 : SUCCESS in 1h 27m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/03ec726d1d494589934ffed0657254cf : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/769dc54f0b3f4d3dacce204a465845c7 : SUCCESS in 43m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ae2078c57b97441ba916e71bcabefef9 : SUCCESS in 51m 39s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/15df4ff0c63c4338942455ad3a295946 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/00eeec15ff90461dbfe2656923807eb3 : SUCCESS in 1h 16m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2386a548f6f042adbafdb4138d15c049 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a417b1d982584c829e9e597c7b787a42 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78ab47c79c25427996193a410297dad6 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7b2e9c0faf2c4a6190fb27570e687976 : SUCCESS in 4m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ff7756d152f249ceba63297c28e2dd3c : SUCCESS in 4m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10076f317c5749448ee18e5c93b82b03 : SUCCESS in 1h 30m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc45f152b4484310a7148783303eca43 : SUCCESS in 34m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66764244f512419aaedc84690ecc6158 : SUCCESS in 1h 08m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ca65b016beb447bb85f40765c098442 : SUCCESS in 1h 53m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"965bf32031ba815491863908f86526eefa94a511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 11:26:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae74d6a59716cfc8eab3a8779b22ca2522c890f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:23:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/16b0800356474540bc0892d1d5f2660a : SUCCESS in 48m 03s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24b7b8a647744056bce07bd5e71c8d19 : SUCCESS in 51m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecff2848bd3747d8b843c075c5325492 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f66befb885314c439ccc3f7a499187e5 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ef6b96fc182416ca5e38db4a994e505 : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be40285787364f89bfb6561643fa6b9c : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b7bee8512e9c474589b3fd2a9d067a74 : SUCCESS in 4m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8ce59a2d83e4e6f8141417fa03f5802 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"21a48281265fb2c6292f0ba7a25a0fa1256b0b55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:23:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11f54cfb096aba21d35e9860167e8deaaf623af3","revisions":{"11f54cfb096aba21d35e9860167e8deaaf623af3":{"kind":"REWORK","_number":1,"created":"2020-01-13 10:04:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/702180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/702180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/702180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/702180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/702180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/702180/1"}}},"commit":{"parents":[{"commit":"8eef0dc2b66027700d9345c50995ab2a4168c2d0","subject":"Merge \"Stop gzipping logs in post\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eef0dc2b66027700d9345c50995ab2a4168c2d0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-10 09:37:09.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-13 10:04:06.000000000","tz":0},"subject":"CI: Fix symlinks for failed build logs","message":"CI: Fix symlinks for failed build logs\n\nFollow up to I5b5f323471676317a2898875cbbf297082328fcc. Logs are no\nlonger gzipped, so we do not need to modify the 000_FAILED_* symlinks.\n\nChange-Id: I5df42ac0ff91b1b97537a4fbc33fc7970f29d60e\n(cherry picked from commit 63f773f569c7601deab40b32bd62ccde513fa46f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f54cfb096aba21d35e9860167e8deaaf623af3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11f54cfb096aba21d35e9860167e8deaaf623af3"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
