)]}'
{"id":"openstack%2Fkolla~702193","triplet_id":"openstack%2Fkolla~stable%2Fstein~I578f7a6d907d5579180bc64c99c7e6c8edf3afc8","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1856693-stable/stein","hashtags":[],"change_id":"I578f7a6d907d5579180bc64c99c7e6c8edf3afc8","subject":"Fixes rally db commands","status":"MERGED","created":"2020-01-13 11:00:55.000000000","updated":"2020-01-14 12:18:52.000000000","submitted":"2020-01-14 12:17:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702193-1579004229994-0548d2f5","meta_rev_id":"910d69aa891d04dab0c1ee804509bb86e855fc24","_number":702193,"virtual_id_number":702193,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-01-13 21:51:07.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 12:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-13 11:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-13 11:09:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-14 12:14:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-01-14 10:10:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 11:42:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-14 10:10:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-14 12:14:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-14 12:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e9d23f3e02728e663e95941dabe1e1a4a0a8928","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-13 11:00:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c1921b7ac14500a15f153ad5de68428028945a30","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-13 11:09:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06631e0fae4d45c1bea36c8ae3948d5fc8f9f294","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-13 11:42:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c05ab4fac202c0dc17a170ad210f029a0dbc9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 12:40:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/173abe796be044a4a323db01e891f23b : SUCCESS in 40m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7ec96549521e451fbcd4c64cbc909528 : SUCCESS in 45m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b343d526ea00403caf183e850098826a : SUCCESS in 1h 20m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9a53549a27364581ada2112b0de5f26f : SUCCESS in 1h 15m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e3a763f7a6c4a7995afa9bf1e37f28b : SUCCESS in 1h 33m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9ab0decfe98740ca9a306e937e4c7021 : SUCCESS in 56m 53s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b1935f8655174585ba7703cc243e621d : FAILURE in 49m 49s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1e2e2327164a47699a315e2ea03365c4 : SUCCESS in 52m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbe89c56ed6041ea81abf93416226c1b : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b48f7fb7605421fadd5213b2ac3802d : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a310bfd9f33b43b8af1b08f0e22bde3b : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed38d82c6924460a81f39b9a377e657 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cca4473655f430e983e79a440d21d47 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e11c3cb259043508c8780cfc4819ca3 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c6b51546b049a89172d307a09f731a : SUCCESS in 7m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03f7ea324f804e1494d84d8d2c60629f : SUCCESS in 51m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e04a6f19dd8247e68ba07867db68a574 : SUCCESS in 51m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3dff74318df4c20a1ef0015aa32402c : FAILURE in 6m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e129c7df66d4458b81316415550fb91 : FAILURE in 3m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d748736754348f2b85954ae3457e1ca : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7640db74327ec98c40d7dc558a016e8616943b9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-13 13:06:12.000000000","message":"Patch Set 1: -Workflow\n\ninfra pip mirror problem","accounts_in_message":[],"_revision_number":1},{"id":"61b0cb9fc9c532b7e2c50d49ffb4191d97426ad6","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-13 21:51:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77f05dc87d08990935a6dd56b20626032f4f816d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:12:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e78a1e271551468faaf0d37d1da4f15d : SUCCESS in 59m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a815d18edf84e39a3fc71ea76920037 : SUCCESS in 1h 02m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/defc17c8dbdd4326aad1d832271ce1d3 : SUCCESS in 1h 40m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/233826e2d3064ca3b78e83fae516d056 : SUCCESS in 1h 35m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/559449901cea44f0ba3ae57edae5435c : SUCCESS in 2h 18m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/36acac996e594c22be23a0fcff4a334b : SUCCESS in 58m 53s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ba23709d31e346299de95dd56e475c00 : FAILURE in 53m 59s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/11523f53d66c4473a4653f97ec17b00b : SUCCESS in 59m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b3b81fd3b49467ea270e7d6ed5e5474 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81790d73cf2c44ee946cf7e817b7e71d : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/150a884dc41148288e7428024977521b : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1949725a8bb0498dba0c7df505516120 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8754e0e0a8dd47a6bf43c9bbadacafa8 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b82a4bf01b40d7a5ba9b48f1b0f030 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73d13fd3bebe430e9823e917f02ffe6b : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0be14792d89a47c3926e81b3f9e41ada : SUCCESS in 38m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/537bc607705d429fb190961913d4f740 : SUCCESS in 42m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c020d5da25445dcb612d82c80d42253 : SUCCESS in 58m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1fdfaeb190014f39b4964ffb2b3ae474 : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/938231a3c2944427bd4cf11d9dea012c : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"32a3ddceb4f18a52959e9ef5da0b70fa7664a9a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 10:10:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b6c3ae1d4c41c02c7bc75c927249c39a6b97adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:10:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f892d5436b82b43cc6090c336fbd36a6250824c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 12:14:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93a113ce151c37948aa194264476595825171973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:17:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf40044f70399ddfc046b48cb4ca58a92d33fc97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:17:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ef504d14e9ee4c57b2b4ce66a53e631e : SUCCESS in 48m 32s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/03e7c5d0fcbc48f9b54288358a658507 : SUCCESS in 55m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c10e81def6f14b8fb3864adae5ecd4f9 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fea1e82c08e469ca073a06faec2e0f8 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f76607eec32a4ce8ada1bb5c9df990c0 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e9ea5b1dc314e828ff4002acdf34742 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2bd9c91cd064224b0a7883b1e6236ff : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe3bad918214315833828128b3396ca : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dd16bc625554408977f6bf52d8410dc : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e44eb4230fef43dc8d1d18afcf66a78e : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bcf60100e312a482c1327a0f3fc30004ad3efa35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:18:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef011f50b3534cacaf3f05bbdfb13f30 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8781fcd56d1542f4b1edf5e9bc823266 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b9485c6f3578a277910c6be6c2963c427f13613","revisions":{"0b9485c6f3578a277910c6be6c2963c427f13613":{"kind":"REWORK","_number":1,"created":"2020-01-13 11:00:55.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/93/702193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/702193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/702193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/702193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/702193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/702193/1"}}},"commit":{"parents":[{"commit":"15c8c73c08b60b77edfbad0ee523c33ca0dac6c7","subject":"Merge \"Stop gzipping logs in post\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15c8c73c08b60b77edfbad0ee523c33ca0dac6c7"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-13 00:36:40.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-13 11:00:55.000000000","tz":0},"subject":"Fixes rally db commands","message":"Fixes rally db commands\n\n\u0027rally-manage db\u0027 command was deprecated since 0.10.0 [1]. Should use\n\u0027rally db\u0027 instead.\n\n[1] https://rally.readthedocs.io/en/latest/project_info/release_notes/archive/v0.10.0.html#command-line-interface\n\nCloses-Bug: #1856693\n\nChange-Id: I578f7a6d907d5579180bc64c99c7e6c8edf3afc8\n(cherry picked from commit 1114859b5cb940290f405cd64e34c6638dcebcae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b9485c6f3578a277910c6be6c2963c427f13613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b9485c6f3578a277910c6be6c2963c427f13613"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
