)]}'
{"id":"openstack%2Fkolla~702287","triplet_id":"openstack%2Fkolla~master~I835d10bfe3ac0ea7382387a27b222849afb84aa3","project":"openstack/kolla","branch":"master","topic":"upgrade_kafka","hashtags":[],"change_id":"I835d10bfe3ac0ea7382387a27b222849afb84aa3","subject":"Upgrade Kafka to 2.0.1 release","status":"MERGED","created":"2020-01-13 20:02:13.000000000","updated":"2020-01-15 16:53:43.000000000","submitted":"2020-01-15 16:50:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702287-1579107009239-71032318","meta_rev_id":"c4c1f6475827173c4192bc0eafc53f586826effb","_number":702287,"virtual_id_number":702287,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 16:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-15 13:31:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-01-15 13:06:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-14 16:39:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-01-15 15:05:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 16:39:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-15 13:31:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-15 15:05:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-15 16:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42fc7121a520871aff53a1ff17646e5a532613ec","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-13 20:02:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"317fe8f81e65d0c329957f1f9c4333b987383888","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-13 20:27:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/427c62b104294e2299597d9074111ace : SUCCESS in 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1e963afcfc92a6dd489862ec2456a6e8fcfe7cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 22:01:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9a8edb9cf16c4bd6a85ac28d429f4831 : SUCCESS in 37m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f0cd5c6618142969f9e30d8dd72e350 : SUCCESS in 48m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7a37751cc59c4be0a3d5ad8d4bcef411 : SUCCESS in 6m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/352accbe06554ba3becfc9156cb94cbe : SUCCESS in 6m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d1d8edb40c574cbaaf2e9488e7ea74a7 : SUCCESS in 1h 31m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/07ec62eab990450db1c4399f5aa75419 : SUCCESS in 1h 21m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7a546bf292864eccbce3c23331f4baba : SUCCESS in 1h 53m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d67d1d03f839481ebb103028f422efb9 : SUCCESS in 58m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a8ca6b9451774fddbe76b09e5b1e358b : SUCCESS in 27m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ea4e9ab0234443e7b7289b85a3dc3567 : SUCCESS in 23m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc559c15f6b240a898628e9f5cbf911e : SUCCESS in 1h 01m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a6d755bf95a412caec74e2f54fc8a13 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bc48a78ea7a4af8a297389630a74e85 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75522d9dd9aa4110b05742f9853a6fcb : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b58fc883bdc146aa8e0bb2bb502dd0f2 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3302ea464eb4369a88c144e72b004aa : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb8daa9b46f54da399654610ecc9c89a : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f7ecbc6007d45c3ac07ccd1505c83fa : SUCCESS in 27m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec4506892695429491e59294b724b797 : SUCCESS in 31m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c204ab6502994d0796f7786e73de6f7a : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c960af64339745f08390d931bd5c2610 : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6803864d4b8402c91a2a3127bbce5a3 : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"716000d141be6b6b784b23491057f8abd9959637","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 12:32:20.000000000","message":"Patch Set 1:\n\n(1 comment)\n\ndo we need to support upgrades?","accounts_in_message":[],"_revision_number":1},{"id":"4d8619c4dd3cd8d3f6e159507ae0ab4a04c20472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:40:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/626f32d15e73402f8bc468281ef2116b : TIMED_OUT in 3h 05m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67dfdb0a901f80906a705867fad2e335ee1e145c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-14 13:51:33.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nIt should be fine to just restart the cluster with the new containers (which will introduce a little downtime). If someone wants to adding rolling update support to Kolla Ansible, that could be done under a separate patch. It is a nice to have for Monasca, but not mandatory.","accounts_in_message":[],"_revision_number":1},{"id":"605d9190c456a60dea0bf8011722f1a345612109","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-14 14:12:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89ddf1363997a91366d2c2650d9f74071be422c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-14 14:46:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/60aad4e96d604247a62b16186b2b2ca3 : SUCCESS in 25m 36s","accounts_in_message":[],"_revision_number":2},{"id":"49c0a9c0282fc362e26cb1ebc9bbb0b0ce2a991d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:13:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/25ea78e3bfc14b54b994391b2c2c7ace : SUCCESS in 32m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b339afa484b1481497ff285cd4dfe68c : SUCCESS in 41m 34s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe5aa497a51049b58277ba45dacd47d3 : SUCCESS in 7m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee6ce3b973e048b3bac72605c5042275 : SUCCESS in 5m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/522f53283e644c01a28cad32624a1054 : SUCCESS in 1h 19m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ed51d7bdb2524416a0a6450cff0f7690 : SUCCESS in 1h 36m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/caa0a9f73c9745ab9adcb9d21641ef89 : FAILURE in 15m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f6552fb70bee4f12add67b0d3c9cd18c : SUCCESS in 49m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d8312f6917147a6ab39b8268c391eef : SUCCESS in 35m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/afaccc5996614754aa2424624a0f7b3b : SUCCESS in 35m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/df36682c06524f75b779ccad2d130974 : FAILURE in 26m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5aab207a22f451baeec80514e00bd82 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87e5d362fd74c0d98b99ad398430954 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70c373768da48fc95457ab2bd29530c : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a237f907d93484f9a1cffb96e644c0e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e7ad9bc129455491170322ffedc67b : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b555636052ca4dabaacfebb60bf349db : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8554e492780490d9c29b9810e58108e : SUCCESS in 37m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c97ea4367e744e1cbb47d0d36a97001b : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16c4841d07104d59893a2ad474424527 : SUCCESS in 58m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/598ab3193a4e4c6bb51d4424725cd9b1 : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7554ab729eff49499e541f2d0a90d1fd : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3aab874a5c6c72c21c5c73e99f601207418073d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 16:30:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32922fe2322dc463235cad7ffa1a8e078438cf8a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-14 16:35:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a36ebce9723a7a4e8b8071fa9accbb3ee572a54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 16:38:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2897acd522567f913d6d61eb556f3334dccccca9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 16:39:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6d017b9c0bef14cff1e03fe4f05b8c70448b3a50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 19:49:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b46772ab7cdc47cba96132678de8764a : TIMED_OUT in 3h 04m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6814ef19f7fd2d3bd91e7a6f59596b7a8c869fc1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-15 13:06:37.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3766e375995e71dc2a82e015fb405cb6f9cf7846","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-15 13:31:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1c5b1b7702c24841811737a1e4d88dac : SUCCESS in 24m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2813a30f878346b9e36c05c9ec514e6a106d4e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 14:47:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/21a30a28764446e18ede28039004ab69 : SUCCESS in 30m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/451d9b5389d441fabd629666cd11280f : SUCCESS in 36m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/61738b450374425481927f9f8d2db225 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3b985e9001b34e72b56c446cc40e27d2 : SUCCESS in 8m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/50c24f26cb2444cf85ef5fcc1ddfe536 : SUCCESS in 1h 31m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/74cde8be1d344b0aaf24421d8fb03d09 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ffa1f1cb4a1641adbdfe38d2163718b8 : SUCCESS in 1h 30m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b8e56029944846a69e2c95b51fb62f21 : SUCCESS in 57m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a50d2ba939f46f6a44320851fb4b152 : SUCCESS in 40m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf6cb9bb41fd4777b65c13e2a7908202 : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/968ee7ec3b3a4f6e83fc4c1abe4521e8 : SUCCESS in 1h 12m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8eefbe96512447fab89a3132e2bf04d : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d1634f48f04850b5c205c08e1e0df3 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9a8aed2ccef41fd9c6a1da065e01aa9 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b7e89363abe45619caf5141bfcd6dca : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1183222ceac4ce7b874bb6aac6f2fa9 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/807e4b6a50fa42ef8a0f99c0daf5f97e : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99df886f732b4681b903a031be232d8b : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/707659d07ac44e03b991a59bf65a0cab : SUCCESS in 35m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f91e4b49f41048e2996e419ba1e95252 : SUCCESS in 59m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cba7f69db2b434fad0b9f27ecf1fd96 : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f1484d1d72143429360f22344338877 : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"413ab2b0150b8b8e35f17e86d7c22ee9dbd9e6bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-15 15:05:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6bf96bf396e2ddeed5ce1d801bdefb1016f42432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:06:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"703ffa56b3a000e338c1e1b314f8facab5786a14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:09:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c127d2f6f06b46a7ad70e955353ad71d : TIMED_OUT in 3h 02m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b77c4d8325eb5362462bd7c25368fb4c3016a12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:50:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"01deb51ae2be73d6c245235c972b6446a46b5b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:50:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a88a82b7eaed4c27b245d592cd111e7e : SUCCESS in 35m 06s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b274859a2754747ab06b4e62dde29d5 : SUCCESS in 6m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/efc4086eba14494c93a4b66d30984f45 : SUCCESS in 1h 22m 45s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/81cda7358caf4eb7a4420ecd71fd8a1f : SUCCESS in 1h 38m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/77dc4ef252b544388da2e01a940fd154 : SUCCESS in 53m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37df5511bda54dbb830b9a0d5a07f446 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c724d593d8334b44a664326479c6db37 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e209555f32ce411cb9fe918594344276 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c51653f808f345bfba5f8f8d45a811a1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c1335254c5488e9aa0e46df918e41b : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f76d544efb940c5a2d2494e39c55844 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8112ae5520804e3bb8657677834b40d1 : SUCCESS in 38m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0444e4c1b7534ef484ac30929ca5babf : SUCCESS in 59m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d4010d67fd5491982d65368f6fa93a2 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":2},{"id":"bb9da2df34f676d7b63de52bc009097474dcc66b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 16:53:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3190ca12b8e047a6a0bdc68211288fd1 : SUCCESS in 2m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d5e56167aa742ebbd17248970c7ddb1 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2a17c0097c2eb4434681fd7ac75173bf3470401","revisions":{"8496db886a5cf97404cbc83da534a7795474e256":{"kind":"REWORK","_number":1,"created":"2020-01-13 20:02:13.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/87/702287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/702287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/702287/1"}}},"commit":{"parents":[{"commit":"e2f013357b850a433a300d4363e764ea9c3b9e5d","subject":"Merge \"Fixes rally db commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2f013357b850a433a300d4363e764ea9c3b9e5d"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-01-13 10:20:57.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-01-13 20:01:36.000000000","tz":0},"subject":"Upgrade Kafka to 2.0.1 release","message":"Upgrade Kafka to 2.0.1 release\n\nWe upgrade to 2.0.1 and not a newer version because Monasca currently\nsupports 2.0.1.\n\nThis also brings the Kafka release inline with the current Zookeeper\nversion for the 5.0.x Confluent platform [1].\n\n[1] https://docs.confluent.io/current/installation/versions-interoperability.html\n\nChange-Id: I835d10bfe3ac0ea7382387a27b222849afb84aa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8496db886a5cf97404cbc83da534a7795474e256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8496db886a5cf97404cbc83da534a7795474e256"}]},"branch":"refs/heads/master"},"d2a17c0097c2eb4434681fd7ac75173bf3470401":{"kind":"REWORK","_number":2,"created":"2020-01-14 14:12:14.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/87/702287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/702287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/702287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/702287/2"}}},"commit":{"parents":[{"commit":"e2f013357b850a433a300d4363e764ea9c3b9e5d","subject":"Merge \"Fixes rally db commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2f013357b850a433a300d4363e764ea9c3b9e5d"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-01-13 10:20:57.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-01-14 14:12:02.000000000","tz":0},"subject":"Upgrade Kafka to 2.0.1 release","message":"Upgrade Kafka to 2.0.1 release\n\nWe upgrade to 2.0.1 and not a newer version because Monasca currently\nsupports 2.0.1.\n\nThis also brings the Kafka release inline with the current Zookeeper\nversion for the 5.0.x Confluent platform [1].\n\n[1] https://docs.confluent.io/current/installation/versions-interoperability.html\n\nChange-Id: I835d10bfe3ac0ea7382387a27b222849afb84aa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a17c0097c2eb4434681fd7ac75173bf3470401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2a17c0097c2eb4434681fd7ac75173bf3470401"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
