)]}'
{"id":"openstack%2Fkolla~702370","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5","project":"openstack/kolla","branch":"stable/train","topic":"fetch_fernet_python3-stable/train","hashtags":[],"change_id":"I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5","subject":"Adapt fetch-fernet-tokens script to Python 3","status":"MERGED","created":"2020-01-14 08:31:20.000000000","updated":"2020-03-06 17:47:34.000000000","submitted":"2020-01-14 12:38:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702370-1579005484387-d94f54cc","meta_rev_id":"c9cc4f29452bcd121ab79b25f2ba6b5d5e8f4651","_number":702370,"virtual_id_number":702370,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 12:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-01-14 10:06:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-14 08:37:32.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-14 12:22:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-01-14 10:06:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 08:37:32.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-14 10:06:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-14 12:22:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-14 12:38:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22c22f9b5593aa4d4ee08d3b0b99414e8c75b97a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-14 08:31:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a863f45e743a0441870d607129cd0b42c7335867","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-14 08:37:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c0970a4499a2d4bd29eed40f8f9a9538fd87108","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 10:06:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb01960c20858113fb6f286effe5768f22d62ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:49:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bb8d13e12a7642c0b53d39b9e24f3c26 : SUCCESS in 57m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b364ae8457a441679ac5ed7e406741ec : SUCCESS in 56m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0dffe71166f64ddbbc4fd2183e2114a4 : SUCCESS in 1h 23m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/41af27b45f824b5c8592f81d0b06f958 : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4a39e7cd5c42413d985aeb4b31962e25 : SUCCESS in 1h 38m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d360f085552e46a9a960aff5c3ecb714 : SUCCESS in 54m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80acb1b61f864284a02fcf945e1dca49 : SUCCESS in 31m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1e6286f0885642a99c0691374333efed : SUCCESS in 28m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e4fec1c11dc043a48a61ef762c7044c1 : SUCCESS in 1h 14m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9c2a2e35c514c4bb550c7ce4146c7d7 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074bb7a8c8e7440f87627cb8f4014b9e : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e080a3f3d584a019bf71622656a0d77 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/904e0563ee9249bfaad299524b553337 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0296ccce71d2466f8be347eda6697ae6 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edda2ffe9346475ba74ad35647c75749 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b518557f71a14f25bbf6946d10482497 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c2cc042f87f49fda669a141f3b88fba : SUCCESS in 34m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/187090c86f1f49228f0757b6c7b18206 : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2274939971ee453aa93f662a935cbd84 : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6006ef5a4c1a4f1baa6493a2aa9de86b : SUCCESS in 53m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76d7cb8243654825851c1a2674189e2c : SUCCESS in 1h 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b1f7b5ed48dc2742cfc48bd5b7e629fd57fc88df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 10:49:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a3a7557c832661b4ee38289f2f06eba176480d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-14 12:22:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26fd33796982765e2bb8c60e7ba20e101e07e54b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:38:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6379232625688e54fbd656b52842941c5fe7e63b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:38:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b3ae4d1ac07849bd8655712fe9634aff : SUCCESS in 56m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bc1bcbe71aba4ab8b4405ae406733cf2 : SUCCESS in 1h 41m 29s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/10350ca083944bd58430203f2a92a196 : SUCCESS in 1h 43m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0c5892ed4ab4437ea2e41990210c06e0 : SUCCESS in 54m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6087a388062a47cfa667d327e229a003 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6981f933174801bb32186ab644ba1d : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b93d4909fea456bbea82d774386fdd4 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5b0a13d26bc44749ed8740d0e194404 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d420be5eb6bc401498e9edf598b8b2ba : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d75cad016004786bfde2ee08d38fefc : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63ff35284be247528ef6b5c6d4a34fba : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77afb2cec2e8459bbef3ac1c1fca8459 : SUCCESS in 28m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3a87386ba8348358d3b5c81f678af13 : SUCCESS in 1h 09m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/48625ae2a4de41af8887e8a63a330b78 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"dffd25b296684d382284d339dd7c66bc63d76e01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65859898f435455b9c2aa552987fc086 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7eaa8f4de22409fa2a6acfe92a3002a : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a457293dc998dbfd6b6126ae2854425d04e27268","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:47:34.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a83173da818e8a4b14812dd111f017b34beb6abc","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"902f16cb1a8efc44da8669a6707d93d47160ae9d","revisions":{"902f16cb1a8efc44da8669a6707d93d47160ae9d":{"kind":"REWORK","_number":1,"created":"2020-01-14 08:31:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/702370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/702370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/702370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/702370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/702370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/702370/1"}}},"commit":{"parents":[{"commit":"1decff518df82e5d104ec929df0219722785c89a","subject":"Fixes rally db commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1decff518df82e5d104ec929df0219722785c89a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-09 17:32:32.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-14 08:31:20.000000000","tz":0},"subject":"Adapt fetch-fernet-tokens script to Python 3","message":"Adapt fetch-fernet-tokens script to Python 3\n\nBackport: train\nCloses-Bug: #1859047\nChange-Id: I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/902f16cb1a8efc44da8669a6707d93d47160ae9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/902f16cb1a8efc44da8669a6707d93d47160ae9d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
