)]}'
{"id":"openstack%2Fkolla~702420","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ie9a0109bd4eef337358d714e4b5088b8e3de2f23","project":"openstack/kolla","branch":"stable/train","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Ie9a0109bd4eef337358d714e4b5088b8e3de2f23","subject":"CentOS 8: Add zuul build and publish jobs","status":"MERGED","created":"2020-01-14 13:31:40.000000000","updated":"2020-02-21 09:15:20.000000000","submitted":"2020-02-21 09:12:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702420-1582276349145-86a07e3f","meta_rev_id":"b72d84ceda369be60ab41ba2187fac10f0837355","_number":702420,"virtual_id_number":702420,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 09:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-02-20 16:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-20 17:08:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-20 17:08:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-02-20 16:37:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 16:38:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-20 17:08:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-21 09:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90082d11f28d0bf0d975e64f60b59bd73a3ef672","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 13:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed815a0165c8bf606b523f7d9c3d0fafc532bef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 13:32:22.000000000","message":"Patch Set 1: Workflow-1\n\nBlocking CentOS 8 in train for now.","accounts_in_message":[],"_revision_number":1},{"id":"ccad473009a75839394e47ea716b75f093a3faa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 15:27:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/14d8f2ea5fff4bcab7fe99314a74e73c : SUCCESS in 54m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/97e2c320420c4d2897eefd934dcd6cd8 : SUCCESS in 56m 34s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2f89b454c549404989f61fa2c8a9c283 : SUCCESS in 5m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d8a32f6efd24630b630ce08c178e86d : SUCCESS in 5m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc6f229387994be79e3a5fb91096298c : SUCCESS in 1h 27m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d2f0889a52ba430181155a31c0d9dc64 : SUCCESS in 1h 24m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e7c57c8ecc8644acb855b920ddbc0514 : SUCCESS in 1h 25m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f287ff904b614f19a23dad1460df57c7 : SUCCESS in 54m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0c954f4417834958b28864ea1080164a : SUCCESS in 29m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c5ddfb87b0dc4e198c0ed4fe0947ee5a : SUCCESS in 30m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8948359e8ccc40619d0af5f4753e51d8 : SUCCESS in 1h 02m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/292ad35a83e1403384f386b0b55ad681 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2033b972f443458b9830a2a17b95756b : SUCCESS in 8m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d58482d4d8e4e308880ffd3a8a72275 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7bfbd72109847dbb0f2ca4f7de0e9b6 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ff301a086174ed6a3693fa0a0403ad1 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e395a777f184e67988eeaed6b596871 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf6a6c4287a145c0b5bd4c8f7ef3572d : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6b282cba2d449bf918615a4af081cad : SUCCESS in 26m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a324206c1b243d4bf81516573ebd537 : SUCCESS in 28m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8e90beeabf849d782ddb9d7b4b43dfd : SUCCESS in 1h 00m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bec66900bc2c46f4bdd890b0c4df661e : SUCCESS in 56m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07fee4450fb64062904a187a3b508244 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c75659d972ebff9d3748ac46d4e5f94f096155ec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-19 12:48:04.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"d778155e7bba7a50200e4ec8b1688e42788f87f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 16:37:20.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"90d0c9d422b21f1d49db7a7ac342a4d6aecf61d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-20 16:38:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"494e6775515e229611f8e404a7dfb51a277af07a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-20 17:08:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29699d93a2b6f636771b2bc29226a3253ecef005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:47:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aef4013ebab5f5f2d3d563b18173e62432d3068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 09:12:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d584c065ee4e42c3983615dbe406d60f : SUCCESS in 52m 17s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9495700cb82e45d1913e57107f0336e2 : SUCCESS in 5m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eb7d913e8b7c49de89abbd6045140bfb : SUCCESS in 1h 20m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/085fb305a2684c609d683743a772c539 : SUCCESS in 1h 22m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7667ec7ba35a471687970a0a842e67dc : SUCCESS in 52m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fa2a270676044e89c30259aa2d4c0ec : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2f904bad0d4f8f8dd77ddb02fd102b : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/164f73e9ad2548bea7deb01610727e16 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41747e31b7a64761a134eeecbf70b6a4 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf09723ccf7b4542958f7621e7454735 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986c1d322a734cc1adb007694d76adbe : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e42779c87544af6b04079848f8fdaa5 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/300625c9b3db4e278c346a38e831dd3f : SUCCESS in 30m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d80ca131fd3b4381963932140a03af69 : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a6d4be444fb40f38ed12c0724732625 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4e8ba1bc83848079a5997f63b0182acf7f08e77a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 09:12:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1903bbed852357d0ec43d62acaab91f49dbed6fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 09:15:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448d9da398a74490b088ecd5dd4d2345 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56172c04a0f44fdc8acd4d9fcc710e1b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f775700a768a010141fa6c82ea94b2bb10a9fe7d","revisions":{"f775700a768a010141fa6c82ea94b2bb10a9fe7d":{"kind":"REWORK","_number":1,"created":"2020-01-14 13:31:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/702420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/702420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/702420/1"}}},"commit":{"parents":[{"commit":"c12ada91d05521996e44bcbfe392ad45367587af","subject":"CentOS 8: work around mod_ssl certificate issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c12ada91d05521996e44bcbfe392ad45367587af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-09 13:27:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-14 13:18:09.000000000","tz":0},"subject":"CentOS 8: Add zuul build and publish jobs","message":"CentOS 8: Add zuul build and publish jobs\n\nAdds build and publishing jobs for CentOS 8 images:\n\n- kolla-build-centos8-binary\n- kolla-build-centos8-source\n- kolla-publish-centos8-binary\n- kolla-publish-centos8-source\n\nThe \u0027base\u0027 image has been added to the list of unbuildable images for\nCentOS 8 (now separate from the list for CentOS 7). This means that no\nimages will be built - the jobs should essentially be a \u0027noop\u0027.\n\nChange-Id: Ie9a0109bd4eef337358d714e4b5088b8e3de2f23\nPartially-Implements: blueprint centos-rhel-8\n(cherry picked from commit b060623b907348e518bdf1afede0b641be56c415)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f775700a768a010141fa6c82ea94b2bb10a9fe7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f775700a768a010141fa6c82ea94b2bb10a9fe7d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
