)]}'
{"id":"openstack%2Fkolla~702442","triplet_id":"openstack%2Fkolla~master~I1af17edc9671237d270c5c6fe70e4420199bdb02","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I1af17edc9671237d270c5c6fe70e4420199bdb02","subject":"Debian/source: do not force tag in build jobs","status":"MERGED","created":"2020-01-14 14:28:03.000000000","updated":"2020-01-18 19:09:03.000000000","submitted":"2020-01-15 11:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702442-1579087192930-9981773a","meta_rev_id":"2bae46a5da8d60e51c128c14dd71b0926cfef3bd","_number":702442,"virtual_id_number":702442,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 11:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-01-15 09:33:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-15 09:46:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-15 09:46:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-01-15 09:33:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 09:33:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-15 09:46:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-15 11:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e70c26b0b18876de7fce19c7d4658640116f18c0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-14 14:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8fea64650bdf9055ba95899cfdf332ef493b658","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 15:05:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e566c20ca3262ede40eda70c483f0dfe4e74056e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:18:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ecefd984009642cc9a11f5510c197adc : SUCCESS in 50m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/61efec2df3354f23aa554cc4e45903a4 : SUCCESS in 39m 38s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b90dd309e2384b5bb5d592fc8930a8c9 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a53f8b2b5e304310924672d973951738 : SUCCESS in 6m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a82f5e8078b741bba2027e2bb75882a4 : SUCCESS in 1h 28m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3783079012bd45b298aa7f5626a6fad8 : SUCCESS in 1h 22m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d917bbf3ffa34096ab8800b224cae036 : SUCCESS in 1h 30m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc287b41fcb64396bb7dbbf9a487f7a1 : SUCCESS in 42m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8c6f167893b4bd99e22f6e66a7baa66 : SUCCESS in 33m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/202a32cb08a348f79c5d067040334a83 : SUCCESS in 1h 03m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/202c1a8b41b54ca4813e51c25c5e82e2 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425233204b56470eb2564a71625e9a3f : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faeda8b2b8fe41fbb92d064ffe32baa0 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28bbe7792bd64c54a0f578f7d27847ea : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e57700f5794030b8aa416bff644cf5 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aaa284dfa2a34cb58cf9b41832bd6882 : SUCCESS in 25m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc847cbef11e4b0e97e94d29cb8d3d38 : SUCCESS in 28m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60581a49d9ae457c9073978ee5d1461c : SUCCESS in 57m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28550daf408e4104838f43c9a87fef6e : SUCCESS in 57m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3dee4d5f0333478cbb2923d817ae8e34 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"edb5092a0460082b2e653e7eefc2e89dd6358cf9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-14 16:21:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddc7ae475b57aaa7576f3e33ecb642ba8ecd8c25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 20:01:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3643bf29efe649d7af21461890c0ad5d : TIMED_OUT in 3h 03m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a6fcbef99857c39aff101ef92aa32b605fce337","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-15 09:33:55.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\nVerified that parentheses not required.","accounts_in_message":[],"_revision_number":1},{"id":"f3c1533c7283419d4a82131774e756e57b7b1270","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-15 09:38:38.000000000","message":"Topic bp/centos-rhel-8 removed","accounts_in_message":[],"_revision_number":1},{"id":"e4456f6ef9d06926240bbe5adfc2f245a7b205d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 09:46:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5a463603b5024efb69dafb1dee18c8b9d603e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 09:46:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36136dc4cb90e3723ea26bf5edd4ecfc73cc68b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:19:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/401b893678c24db2b49376e8d7b1851b : SUCCESS in 49m 42s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8598a56fd98743cf8d0d29d1a20e79c1 : SUCCESS in 7m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca39b149c34b40f4baa5be09b02edb49 : SUCCESS in 1h 29m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89f3a2b6dee549e9a15332ff09440b62 : SUCCESS in 1h 32m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6908b8236ef64a82b1b436228f478c14 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43eb190490c84ec8bd2234c2b2f094d8 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bdad71761e64f6c90e75b27c8aeb4a1 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f9741a4ab0e43dbbd5b5d6deb3e0d2a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ee5305881a427a8a8ff78ee0f655e8 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/34b10047f17f4572b6cb242dfef11fb4 : SUCCESS in 33m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22b49fa2c07f4a80aecf881cb665901e : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e9b06f37d8d4220a1797e5355beecc8 : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ae082d51404d266f7212d2ba15910c203b0c6e97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:19:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1e9848df74a86f78ba2f044fd9c6a0ed37e20390","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83aeef7d2ed446409d9a0a532464545e : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2bae46a5da8d60e51c128c14dd71b0926cfef3bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:09:03.000000000","message":"Created a revert of this change as I02932857454bdb09b72f539038d394760457109d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca4da16836e4b3ef1d84059a63378c3f5bb2517e","revisions":{"ca4da16836e4b3ef1d84059a63378c3f5bb2517e":{"kind":"REWORK","_number":1,"created":"2020-01-14 14:28:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/702442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/702442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/702442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/702442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/702442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/702442/1"}}},"commit":{"parents":[{"commit":"f91ddee1c18cf1000f190396ecee671708b88bdd","subject":"Merge \"CentOS 8: Don\u0027t force tag in build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f91ddee1c18cf1000f190396ecee671708b88bdd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-14 12:33:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-14 14:27:53.000000000","tz":60},"subject":"Debian/source: do not force tag in build jobs","message":"Debian/source: do not force tag in build jobs\n\nMultiarch publisher uses \u0027branch-arch\u0027 so set it in kolla-build.conf\nfile.\n\nChange-Id: I1af17edc9671237d270c5c6fe70e4420199bdb02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca4da16836e4b3ef1d84059a63378c3f5bb2517e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca4da16836e4b3ef1d84059a63378c3f5bb2517e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
