)]}'
{"id":"openstack%2Fkolla~702620","triplet_id":"openstack%2Fkolla~master~Idddf51730821644948f51577807f5beaa18c6262","project":"openstack/kolla","branch":"master","topic":"ci-unb0rk-failed-logs","hashtags":[],"change_id":"Idddf51730821644948f51577807f5beaa18c6262","subject":"CI: Fix symlinks to failed logs","status":"MERGED","created":"2020-01-15 09:56:09.000000000","updated":"2020-01-20 19:22:25.000000000","submitted":"2020-01-20 15:27:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702620-1579534026892-641ca554","meta_rev_id":"b1b29e38c7b4e907b7404496fb90a57f744aa2ea","_number":702620,"virtual_id_number":702620,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 15:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-20 09:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-20 11:40:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-01-20 11:40:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-20 09:30:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 09:32:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-20 11:40:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-20 15:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa94162007ad5dc4bb32d09a3ee13466917ec017","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 09:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09c079e63eb257491c17a9e9fd60f6b6b35e5280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 12:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/54c666b6f8bc449ea4bc991b30788802 : SUCCESS in 30m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9b538660467d45ae8102de650fef43c0 : SUCCESS in 34m 29s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/61793a81ccc542ac94c3a7c070d823da : SUCCESS in 6m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6cc0a1e1867846bc885f8fb390c26e38 : SUCCESS in 7m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e6db86842d354c789f1aa72d4ff44a51 : SUCCESS in 1h 21m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/87095c6f856c48879d32312373dcbee5 : SUCCESS in 1h 34m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4312e9bdac534470a5252b73d2e77060 : SUCCESS in 2h 05m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ef947d9533a46b4b20b09258df23550 : SUCCESS in 36m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/82bf310e95f540ed9380ae1da8ab3183 : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f296963a4f44b0ea1f48d88da8682c5 : FAILURE in 32m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a92f0c0fa4ec4160ac5de46af8636fd0 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f538d52470474299d6f2be141d4f7e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/862c2f6ff801492da0a37d40c1e0e1b3 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4589d95a7a88400d87b230a1f6434c6a : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc0776a4d264e10b96084aad10a28c3 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cff687ecb4744322b5674f176a98e734 : SUCCESS in 33m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0894ad0304b4ff1864b131409691f63 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f830f0141ed4499bbfbb1bac569f49d8 : SUCCESS in 53m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d67289e29edb4b3f871cc467fec63cbe : SUCCESS in 52m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5ba8ee99b534ca189f2a14a068a57b0 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"61ad07d4f56d508f3fa04e6218abbc729de4e118","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 12:55:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5fd25603982482d8639c511ed8462b2 : FAILURE in 1h 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5af6e8c1f03f4aaaf93c642c241f730554962053","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-15 13:02:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5e123e0766096151c1f30afdeac0aa6a73dfdf7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-20 09:19:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3df7e7b3112e3daed251da7b5a8979ad1784e5fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:30:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"902f9472ad477dc0c4c8ad39de58105821b47ba5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:30:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fb5fd9240be2afd9bf08ba99baf6cdf5343f0f93","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:30:36.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"e4029f188e97f3d96a8380d2f9bb75ef1860cf70","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-20 09:32:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c21bd111de45039f34a0eb9f6fc0ec7c9dc9fe7b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-20 11:40:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c50ee89038d199f148226901b3ca05f56722be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:42:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/abea43f5def8452397b34f18ca1ff0c0 : SUCCESS in 34m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c1703be174c74ff88133864cbf61a22d : SUCCESS in 55m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f4c4757888f04ba3a9846f6472296e99 : SUCCESS in 5m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b32b22c800e444a9a1f86902835be321 : SUCCESS in 8m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/69aaeeedf27b4a88981948086230b94f : SUCCESS in 1h 24m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eb19385453dc4961999b518baaf18974 : SUCCESS in 1h 42m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2e50eb8f4808438d89eeefdd24367c17 : SUCCESS in 2h 00m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ae33d35034a4a0bb3a668a087c0ce9f : FAILURE in 28m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a2fa6c731ac43d48b5ceaffdef92a6a : FAILURE in 37m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3760f4b1807747fd8ddcde1a81973924 : SUCCESS in 1h 20m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4fea959a4524294908a5affc7b9f55f : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d2fa40916c423098d78ecf80b67dba : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b5889bd3944793a088bd2bdbd5f95e : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f1242dc37c743aab7f34605665f5987 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c809e2a684f34ba8b30523cd66c49eb9 : SUCCESS in 7m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f631f4e9a2b94145b2cf51e087e5ffbe : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/06417fa51443417aa8d3052f710fc268 : SUCCESS in 36m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78022a2c8bd94d0aba6bf62668e33528 : SUCCESS in 56m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cbfcc1cea364b659b0c40e89b12a875 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9760211b843a43569ee92010f57c9edb : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":3},{"id":"73600c25bca33c8bddf743f40dde1be2fa840a2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:36:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43056ebc52a8481ca7fdf0a85754de65 : TIMED_OUT in 3h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f653de0fbbc6265e7937adb90ec103c3a448f83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:38:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"13d865a0ddc91022e61af7f48345f0590a590c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:27:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/508330836e274db3ba9d4c48dcd722b6 : SUCCESS in 39m 25s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/14f172bee2544d09b8b5493ecab7d97b : SUCCESS in 6m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/56f619b13cef43f8b14fc9953ed305b2 : SUCCESS in 1h 21m 58s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cdaebfd6d5dc45ef9d2c75dd93f1bb18 : SUCCESS in 1h 43m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8013856c52b041f5b59b717ee6b6bec3 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a4073cf6ef423f8c0a2d7d1531634e : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3c14ba27a741f18b1c438313ead78b : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de3de8e7c1144b35b57fcc618f770b3f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5edaa00f2c4bcabc34b835936a8587 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b33d32aca58a4d29a42b385814f14640 : SUCCESS in 29m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2be692dd39ee410eb8fadb58ff7b041c : SUCCESS in 53m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b630c62d650b4efead14e3069491f29b : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6d7d807415a51e44fad12fd0386e869f77ac3899","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:27:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fe1abd099d66669d64e252423057fbdddc3510b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:30:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecff9a7284545d5ae7d75534193d12a : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"7461acccd0af4507fde1f602735c40a3f2c0062b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 19:22:25.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a74a9408b85443b74b208fddf2c2b09937fd8b2c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a0182f15acbabc88df0aa660abce81d61f528da","revisions":{"f23c7db7294d5beaf3b6395a0d099ab54c851812":{"kind":"REWORK","_number":1,"created":"2020-01-15 09:56:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/702620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/702620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/702620/1"}}},"commit":{"parents":[{"commit":"58f7c1f1e436ac9b94e7a248b12ecd123b044fee","subject":"Merge \"Adds monasca-ui to horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58f7c1f1e436ac9b94e7a248b12ecd123b044fee"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:54:18.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:54:18.000000000","tz":60},"subject":"CI: Unb0rk failed logs","message":"CI: Unb0rk failed logs\n\nStop renaming files. Swift should be able to serve text/plain files\nbased on content, not extension.\n\nFailed logs were b0rken because we stopped gzipping log files and\nimproperly rewrote the logic there.\n\nChange-Id: Idddf51730821644948f51577807f5beaa18c6262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f23c7db7294d5beaf3b6395a0d099ab54c851812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f23c7db7294d5beaf3b6395a0d099ab54c851812"}]},"branch":"refs/heads/master"},"e145bd50537f433f918e62534a6f64458300dcb1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-20 09:30:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/702620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/702620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/702620/2"}}},"commit":{"parents":[{"commit":"58f7c1f1e436ac9b94e7a248b12ecd123b044fee","subject":"Merge \"Adds monasca-ui to horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58f7c1f1e436ac9b94e7a248b12ecd123b044fee"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:54:18.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:30:16.000000000","tz":0},"subject":"CI: Fix symlinks to failed logs","message":"CI: Fix symlinks to failed logs\n\nStop renaming files. Swift should be able to serve text/plain files\nbased on content, not extension.\n\nFailed logs symlinks were broken because we stopped gzipping log\nfiles and improperly rewrote the logic there.\n\nChange-Id: Idddf51730821644948f51577807f5beaa18c6262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e145bd50537f433f918e62534a6f64458300dcb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e145bd50537f433f918e62534a6f64458300dcb1"}]},"branch":"refs/heads/master"},"4a0182f15acbabc88df0aa660abce81d61f528da":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-20 09:30:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/20/702620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/702620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/702620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/702620/3"}}},"commit":{"parents":[{"commit":"1cb4760b055bcc38380e6741273337c33c0b563c","subject":"Revert \"Debian/source: do not force tag in build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cb4760b055bcc38380e6741273337c33c0b563c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:54:18.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:30:27.000000000","tz":0},"subject":"CI: Fix symlinks to failed logs","message":"CI: Fix symlinks to failed logs\n\nStop renaming files. Swift should be able to serve text/plain files\nbased on content, not extension.\n\nFailed logs symlinks were broken because we stopped gzipping log\nfiles and improperly rewrote the logic there.\n\nChange-Id: Idddf51730821644948f51577807f5beaa18c6262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a0182f15acbabc88df0aa660abce81d61f528da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a0182f15acbabc88df0aa660abce81d61f528da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
