)]}'
{"id":"openstack%2Fkolla~702622","triplet_id":"openstack%2Fkolla~master~Ia6a5f985b72127b426ce94334e04725983ab5fdd","project":"openstack/kolla","branch":"master","topic":"ci-unb0rk-failed-logs","hashtags":[],"change_id":"Ia6a5f985b72127b426ce94334e04725983ab5fdd","subject":"DNM: breaker","status":"ABANDONED","created":"2020-01-15 09:59:04.000000000","updated":"2020-01-15 13:06:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79ceaa6d87b6aa97d143fcc05dda7cefab060225","_number":702622,"virtual_id_number":702622,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-15 11:10:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-15 10:26:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 10:26:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-15 11:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02c087712a27877fadd61955a8987f368cb45c22","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 09:59:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49a46ac21368f93bfa1b883753ccc60f0da384e0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-15 10:26:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/eaf846af07d14d6b94e0983d0acf9aa1 : FAILURE in 8m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f4884ab026942d83d1c608a496124d7541c1c6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:10:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a4718479b0c347e88617162701ca58e4 : FAILURE in 6m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ee41dbd4c6e641cd8ce6880058f383d5 : FAILURE in 6m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1bfef68938d7412289bfa227b9e798bd : SUCCESS in 5m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce33ffbddeff4a8890a34c05156e85dc : SUCCESS in 5m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6f804a404a4f4c2fa4cc19b1a597aae4 : FAILURE in 10m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dba025a2420f42bbb1eac9f92765098b : FAILURE in 10m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e9954b4807f2481aa0fc9bd7fac19925 : FAILURE in 10m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/95e38ebca6034aa69d8a04ea9375dae0 : FAILURE in 17m 30s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b7be32c7d2574aa1a7df3122d182997b : FAILURE in 10m 01s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/576859de20974d93a3fdb237ff88f03a : FAILURE in 12m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ac85f6ed84d5458e8df3e1a28de203f6 : FAILURE in 13m 40s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3d7b9c75534b4bc4b64652048991e1c6 : FAILURE in 14m 23s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f6598919fae54df3a08bb74a30f83d9f : FAILURE in 12m 15s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4ec91b5baf174692b7ce18cb1df2df26 : FAILURE in 10m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f15c987f6c9f43a2809e813e0e28dddc : FAILURE in 4m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41378ae17d384ee692d18d993071af38 : FAILURE in 3m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f5c7e660fec845c7a63b56921b642d2b : FAILURE in 8m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae7757a35fd40ea9359d9e54597746d : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87f085084d94462ab3685111c8d8c82 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dde0b00dcac45b288df0b4300723bb5 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90b8d949bab34cbfad98c81a863e3d31 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5618308fda441879793026bd2cb7bab : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb2e0b163eef4a888659bd60bc18435b : FAILURE in 6m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cde9606eda2449ec94ab85d1c2273b15 : FAILURE in 4m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e40bf42044f485a82b4622b120fd109 : FAILURE in 9m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/edc87af2e57b41419de02f1d6fadecfc : FAILURE in 10m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dad45acc2fc4ae186e52505cd7d7776 : FAILURE in 9m 52s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/3eff7d01ea5c4e6f97679da4e7a0ed0f : FAILURE in 9m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/44b5ae7c70a643a0a22a91c2f8a06c84 : FAILURE in 11m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7a109a276bf742edb9698eafcab4786f : FAILURE in 10m 52s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/26df8cc9dbb34574aed2dbe99718f417 : FAILURE in 11m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/42334bdc529f49799b2bec21071f14dc : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a435275e350feab8703d3143b55b1dbf2bc07531","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 11:25:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/908300d6d3f748acadbe1a38dd356f10 : FAILURE in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"116ccc32e3926dce0b1911b5ec454d3780a28217","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-15 13:06:16.000000000","message":"Abandoned\n\ntoo scary to keep","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"490b71872c6352d9e1cce19cc747e311ccd7a5bd","revisions":{"490b71872c6352d9e1cce19cc747e311ccd7a5bd":{"kind":"REWORK","_number":1,"created":"2020-01-15 09:59:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/702622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/702622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/702622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/702622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/702622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/702622/1"}}},"commit":{"parents":[{"commit":"f23c7db7294d5beaf3b6395a0d099ab54c851812","subject":"CI: Unb0rk failed logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f23c7db7294d5beaf3b6395a0d099ab54c851812"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:58:46.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:58:46.000000000","tz":60},"subject":"DNM: breaker","message":"DNM: breaker\n\nChange-Id: Ia6a5f985b72127b426ce94334e04725983ab5fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/490b71872c6352d9e1cce19cc747e311ccd7a5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/490b71872c6352d9e1cce19cc747e311ccd7a5bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
